)]}'
{"id":"openstack%2Fmistral~930799","triplet_id":"openstack%2Fmistral~master~Ia9dc54a929ea2f69b871d099dd827d06efb22078","project":"openstack/mistral","branch":"master","topic":"remove-eventlet-workaround","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-06 14:23:06.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 12:59:55.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ia9dc54a929ea2f69b871d099dd827d06efb22078","subject":"Remove workaround for eventlet \u003c 0.27.0","status":"MERGED","created":"2024-09-28 05:26:49.000000000","updated":"2024-10-06 14:23:06.000000000","submitted":"2024-10-06 14:23:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930799-remove-eventlet-workaround","meta_rev_id":"1ed7b8e2a4ac606eee38ab4005289fc32442bb1f","_number":930799,"virtual_id_number":930799,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 14:23:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-28 07:30:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-06 12:59:55.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-10-06 13:33:31.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"ee4dc90c5c72d94042f5979d4e1f71f97b99750a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 05:26:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95a383e8da4a49a5d8047c2a179ba499b9d24ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 07:30:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16fadac6cfd34c2aa6c687e25d6ad64b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9f0218d13949f983b0ccc7a6223d71 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b330549f04d441d8dfaa7cbc6320869 : SUCCESS in 12m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f271f4e3b0e4b338beac6adaa11e868 : FAILURE in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a5339627434aaead3f8c50299df100 : SUCCESS in 4m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e65d83b84b0d40c194da4c4e2c98af91 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df82f9b35d1d4a4e95038d832447bbaf : SUCCESS in 16m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2b02fa39d49f436ca8849d6ad4492eca : SUCCESS in 32m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b234379cb2a245e584daa3a2681e0b95 : SUCCESS in 22m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f29001f42d2c4843b60e76cbe60f6400 : SUCCESS in 37m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1bbe2944f6434fa796b84dcf98088cf6 : SUCCESS in 41m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9ade1fa532be4afb865e7db300d1d784 : SUCCESS in 31m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/310d963870734145a59eea034f45b7f9 : SUCCESS in 40m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/282de7d78ce74a0eb7649c771ac8b6b2 : SUCCESS in 39m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0625612617694944af27855d65929886 : TIMED_OUT in 2h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fbf351139689f2c38233f05e183cd2121d6b765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-30 11:00:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"b036e563014f2c6f83c600a0b913b98906040d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 13:03:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b310d06718194b6b9d27a32fdd96a99c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9730bc4edd16472bac327c119f8af608 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05830f7ad088456d879c51f881b99220 : SUCCESS in 12m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e31c264122747d2abf5891dbd071a90 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa968e3ad0b4c7187827c33ccb5f91a : SUCCESS in 7m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf0958a2cc2f4a13886d43faa31ce4ff : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4f8f62fd8a8498e85cd7c4f1ece7840 : SUCCESS in 14m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f1e296169feb4ed48235e47c08b99d76 : SUCCESS in 27m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7acc94836b846d58ee77472848e816e : SUCCESS in 26m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15f82ca378d2463aad669f39a1d16b65 : SUCCESS in 28m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/38c02faf140f44f599c53d2487f364c6 : SUCCESS in 33m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6a448f38747e49df92eda58916395982 : SUCCESS in 32m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/44a9ecd484d0428d9cb750156cfbe670 : SUCCESS in 48m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bc4db10101c64b70b2a718201bbeffe7 : FAILURE in 40m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/efcbb80cec1f4fb3bbe5ed8e6b311eef : TIMED_OUT in 2h 02m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6108c4a53eb18b9ba3277a9db188c9ef75f47805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-01 11:22:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"14f2125356c5748ebe34f6af5710a5cfa4aa548c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:29:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74cd8df33c044867ae477fff83e2471a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba396673edf74d59aa35af9c2f7e3cac : FAILURE in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1856d922285f4ef3a6c5ad4c175342da : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58e244085807432a8e9af0fab89bb454 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe76c0e0453e4f6482ae4ed73e6ff905 : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4f605762bef491dbec1c568b094b204 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8adf87bf0090421b85ab37631235da7c : SUCCESS in 18m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a2e21848798f446bb05ece2bde8cc6a5 : SUCCESS in 26m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e48fa78feaa4816a707e1fedc272fdf : SUCCESS in 40m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04c3c04e75ff492dab882e8a89473e23 : SUCCESS in 40m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ba69d9a7a602439bbea9ba2c306d1e8d : SUCCESS in 22m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c59110a822634508a775cf4392ae2d90 : SUCCESS in 34m 13s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/b653ae8d8d614b4cb2ab8549201282f0 : FAILURE in 44m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/97f94ef63d0c46a1ba179a2396115ed9 : FAILURE in 55m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d0bea70af3f4db4b1413e8663c0d21b : FAILURE in 50m 17s\n\nWarning:\n  Comments left for invalid file mistral/tests/unit/engine/test_error_result.py","accounts_in_message":[],"_revision_number":3},{"id":"7f0e7d06a625b116cd01852186687b71f9d62258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-01 16:07:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0318d40bddd6a17e333d68ee7e5ff64d30d49ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:16:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45f74473754641f4acddced13d2386f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a88da76169241a99c282204a5732eac : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fe8549d3a3e4c9285a6d7aa813f1129 : SUCCESS in 15m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73a29d2961644935bf46f6907f1e85af : SUCCESS in 13m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a8df75d3854d278f2c7469e3401214 : SUCCESS in 6m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bee627fcd57d449daac2da6182827f4d : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb346803050d484c8a480b470abe3837 : SUCCESS in 14m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/75efe217a06145f083a64ec860b67a14 : SUCCESS in 43m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4264a65244a64c21a1e03217e18f09e7 : SUCCESS in 39m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63f873559b1f45f8953f34fff637b061 : SUCCESS in 37m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/03cb17dfa16447e084185337a980f876 : SUCCESS in 39m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/63f43dfc6f634bc9a2e0182e270deaea : SUCCESS in 36m 07s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/0d43c58961254813b949c9a9ed40e26b : TIMED_OUT in 1h 00m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f1a3169f9b76495fb5dcb65161dacae4 : SUCCESS in 49m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3f11c1e2d815426b97272278c8ee8656 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d5b8a74853cbe8ab49dc1f27945023b186737bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-02 21:19:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"911216542778ebe669913f0a5d35d69edd7056f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:06:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86b996abb77b4847b57c4c76f78184e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cef43e2233348c682a618c85c2007e0 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4b7c17694c84d4da21fdb084a0dbed9 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3634fca946334ce4853bf7d296b14eb3 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4102acef9b94bc6930084f0b9a1b816 : FAILURE in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f692b502503479f9f51aa7f35bd5f4b : FAILURE in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bbdc95d8745481ba4b7d11b07bc4036 : SUCCESS in 15m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b3e503a0f0ef49108e28a86ac8565c9c : SUCCESS in 36m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecee6234a78d4aecb13abf1a8ea527e1 : SUCCESS in 29m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad72ac58be87407b9763bc612ad5568f : SUCCESS in 26m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1c10741f461241c9b93daabb8797cd95 : SUCCESS in 28m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/66b45b42cf884d09bc9e10e8ac494c12 : SUCCESS in 23m 16s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/07ba95fde2ce42c1863f31cdcbd0d10d : SUCCESS in 9m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/486dbbbeefa142e3927dc56baf1a9798 : FAILURE in 12m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/16835f3fae8f48d8bd3f709de9a9dbb6 : FAILURE in 10m 54s","accounts_in_message":[],"_revision_number":5},{"id":"6c6734c38f77a8d8562da27ade4b498685fbde3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-04 21:37:18.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"89d0a232b361ada0e42e2980d36be8526e54055a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:33:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe847e84c2774d00aee3dc3b0f25c18e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e877f36169a34e51890f3ff0ed7f38cf : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9214263713c49c38ff87d6086960a15 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d8f1c4d48524b1fbd4641b800bac093 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec402840d999491c951ed97cc658f0ad : FAILURE in 4m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5b561bc0b894041a7f50a95a10b7a18 : FAILURE in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b3f6da831e54e959cc3266c3e8754e4 : SUCCESS in 15m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9db0d9d624174cb9b8c26b91b648a34e : SUCCESS in 40m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd88023c98e6480a8a6cce7805d8123a : SUCCESS in 27m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/734b50b8443841d1b234c11840478f24 : SUCCESS in 26m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/467e7f55215c44519dac3974fb10198e : SUCCESS in 41m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e4422c1cef244d1b9865db15fdcf5f38 : SUCCESS in 46m 23s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/e8e59cd582fb49529f17d2d848c22780 : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":6},{"id":"85c4316847eb6fa2e3f06022770ba2c9028d5100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 14:31:59.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c9b10cf2e69e4ae7dc18c7631536738c6280b7bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:29:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ee7065f5f274ff4b6f1f3d4e4f143a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c639644c4074a42902f3c80d2333715 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef984ec08c044415be1cc1872e251e39 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7339e31b913a4db1beff132591f5adaf : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35bf49a0b3094dda98ddf49bb7cc02c9 : FAILURE in 7m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3250a8fccbd04c1b97b877440273a83f : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68a7b9bffe8d4e19b52ecf5f43481fca : FAILURE in 16m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cebf8aead06b4f02824f6894a8d0b552 : SUCCESS in 32m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9ba4e5891ca4871aeeb6fba6d03f86a : SUCCESS in 32m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33fb998de8504de8b862f7077a09f572 : SUCCESS in 38m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd5f89d7478e4ae6a149bce44bfe1d09 : SUCCESS in 22m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e2b27e61d374272808675acde59a61d : SUCCESS in 40m 49s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/ef893dacc5e545b9b1ca090680264411 : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":7},{"id":"5c318abe840139da57f65b5ad1d14f92f55e70b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 18:10:33.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"f08c27ff20d914f3a5a683a8ce27c45a1ba1c81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 19:04:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d3dfcba4654efc86769a1ae6732789\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106074023f534bc095d953a392f6467c : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8134d44436894498a489e184ffcf94e5 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed610b3b64134462851c43d3f51e61dd : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9a9e65b7994b88a2e942c5c991ab08 : SUCCESS in 5m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/28cd6c15c5654ec0803d4bc265d6407f : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73057d22afc94a0aa83889b596ed77fe : SUCCESS in 11m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d386abb115d041e982f7853c6c630270 : SUCCESS in 23m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd6a5ad4e83b40ee9fee3cbe37088cd0 : SUCCESS in 25m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8861c7d4fb1c4aa1a46a55e3c240e617 : SUCCESS in 42m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c65678f5d5b41858c309d217922f005 : SUCCESS in 21m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/adabc08d371b46ec9643bde4ac4bc354 : SUCCESS in 25m 47s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/a3ed322975544cb7ab0564240ccfa910 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":8},{"id":"e67020eaa30d58e6cad7c276d75f9411ed0781d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 09:22:28.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"12489c2a74974375c4452095b146f8226786e857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 10:16:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/774e50fa59a14453a50aaf8a70373915\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdef04c72d5a493f85f70309163e2473 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fac0e52e7284b41896b56f976a2e13c : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/965b853d1e894666a4457cb4a0ec3d4f : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb70847e1a143fa9ccb498c338b2779 : SUCCESS in 6m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c94a1806d1b44fc89bee5bc64b6ebe2 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64d20152beb5451faaab935c9ec89567 : SUCCESS in 13m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4c8b3eed43454e319439b5a7a81502ff : SUCCESS in 25m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/478ea9f9d7304a76b2941e802c1ab032 : SUCCESS in 23m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a220aec0b47e4b349fbb5f4b5c5de9c5 : SUCCESS in 36m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c9f19a89328e4e57b3ccdb90a3125095 : SUCCESS in 41m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/111ea947e75e4081a4a73f46bbc7c999 : SUCCESS in 25m 40s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/a3d0b3cfee904962832a540f7c0ed0ec : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":9},{"id":"3cc1f3582aa304bd0ab5f43994444cb4cdf272fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"413a1280b0130ed7783b37a7fbacdd9660b7640f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:02:04.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5aa9328f6f04a4c805fb158a43e384e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b112ff06f0441dac3e9294aefb1491 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb499e28a9849c2833b14d3558b8988 : SUCCESS in 11m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4316baaaca6448088d3051d8d2f8a62 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942965e1f58b4b6780dcbd3a21955551 : SUCCESS in 6m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/782ad86edd28448e8fea50ea89526574 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e4493036e824efc92790da17433146b : FAILURE in 14m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a3fa524b43f24d3d84cc1c6d310e73c9 : SUCCESS in 22m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/097a5a780ec04f9c8aa9d6fc67994e41 : SUCCESS in 26m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd2fcb35a78746b58768fe34cc6cae25 : FAILURE in 18m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e61ac4b3de024cbd8e5c116687a73ec1 : SUCCESS in 29m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d994302f171a4d0aacd9ec25c8fa1f4a : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":10},{"id":"0e31a86f49e17f70c3c786aa51e0318abc2e99d3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 12:59:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"76f69a12af03e919691f59765385cc986f4985b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:26:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5896ec02223e4b8e94aa3a1bdf05efcd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241107d1b86a4df4ad2e14253cf02eb1 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53672a2f4ff94b02a7c8c0c551a8649a : SUCCESS in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f56c9c1c7fde45398d2f22004ebdf1ed : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b887d6c563451ea9a302b9dc3a363d : SUCCESS in 5m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ed5699891a04352a2a69ea857a667bf : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd8998b4681f447fbfbd844ca4573ca2 : SUCCESS in 13m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ad5a5dac8b764d799498bbb18024395e : SUCCESS in 24m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08a9bed8654346a498a66d4b35b490a8 : SUCCESS in 23m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0249a598c7d34d0587419ff5db9428e8 : SUCCESS in 26m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/edda4a590894426c94ca7966442ffb18 : SUCCESS in 24m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/06c595c8ad8c46efb36857dc09794c7b : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":10},{"id":"aed639587149d7b6ef0f3d82325a2c4a6ab18737","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:31.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2925914c6bacd765172c31be1cfe733c1ae7660a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:36.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8671c747f1e8f73d4e14d05e1d225ef9574cf6ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:06.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf4353335ae44817ab52fa7167e12156\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07a8deb757547c7ad22b3478129c318 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ec98a3d94c64c2c9574c97cb2798aa9 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8eaf27e0cd747dcbb9e5b357163be42 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bfae89d51f4dfaa3bdc97d1f8c2eea : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14714de4ed4f49c881338cd39ad1af88 : SUCCESS in 2m 24s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/abf70b07b1774b5ca4bf223d27177974 : SUCCESS in 38m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f50c6fe07d7147e790c9c640f6171f4c : SUCCESS in 39m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/348daaac57fa46699814d3194026f61a : SUCCESS in 30m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e4a476c008a84fc8b5f05ca37aeda37e : SUCCESS in 39m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/36e262dfda0646ee9d3598b61af3db83 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":10},{"id":"1ed7b8e2a4ac606eee38ab4005289fc32442bb1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"ffe7db85fff8d6d1812a2dcbf8e9b6a8c6f7b3ae","revisions":{"0839f056bfa826c6fdfd1037fde601a2999b8b7e":{"kind":"REWORK","_number":1,"created":"2024-09-28 05:26:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/930799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/1"}}},"commit":{"parents":[{"commit":"a3e8f5b0437c2497b1f5e80e50fb873df38f9f53","subject":"Fix some race conditions in mistral CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3e8f5b0437c2497b1f5e80e50fb873df38f9f53"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:40.000000000","tz":540},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0839f056bfa826c6fdfd1037fde601a2999b8b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0839f056bfa826c6fdfd1037fde601a2999b8b7e"}]},"branch":"refs/heads/master"},"152b4b714a5eb22f76cbb68aab4446630b63c156":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-30 11:00:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/2"}}},"commit":{"parents":[{"commit":"183a63b725cc1c57dc151182ccc62f37d4af5a2a","subject":"Fix py312 random int range","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/183a63b725cc1c57dc151182ccc62f37d4af5a2a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 11:00:24.000000000","tz":0},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/152b4b714a5eb22f76cbb68aab4446630b63c156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/152b4b714a5eb22f76cbb68aab4446630b63c156"}]},"branch":"refs/heads/master","description":"Rebase"},"c80fa68f79aa0f5fabb33183f31dcb5de5c4906b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-01 11:22:27.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/3"}}},"commit":{"parents":[{"commit":"cd6e7b2ea8f6e0566fcf585aa9acbc0e26275832","subject":"Add unit-sqlite test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd6e7b2ea8f6e0566fcf585aa9acbc0e26275832"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-01 11:21:54.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c80fa68f79aa0f5fabb33183f31dcb5de5c4906b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c80fa68f79aa0f5fabb33183f31dcb5de5c4906b"}]},"branch":"refs/heads/master"},"02294f1a6c8b74b7a4f2d5f431accc824ce80bbd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-01 16:07:24.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/4"}}},"commit":{"parents":[{"commit":"032a0300b5bc622110c300201f643b593b3e547a","subject":"Add unit-sqlite test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/032a0300b5bc622110c300201f643b593b3e547a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-10-01 16:07:19.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02294f1a6c8b74b7a4f2d5f431accc824ce80bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02294f1a6c8b74b7a4f2d5f431accc824ce80bbd"}]},"branch":"refs/heads/master"},"d49d0d43a72b0b58fd9636342eb31a6ed9228b32":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-02 21:19:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/5"}}},"commit":{"parents":[{"commit":"a4e36894f6443977b35195bf5061f80db193fbed","subject":"Fix some more race conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a4e36894f6443977b35195bf5061f80db193fbed"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 21:17:38.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d49d0d43a72b0b58fd9636342eb31a6ed9228b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d49d0d43a72b0b58fd9636342eb31a6ed9228b32"}]},"branch":"refs/heads/master"},"cbe578f03f09a6d7f43de3d33e238f3934bb2886":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-04 21:37:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/6"}}},"commit":{"parents":[{"commit":"9c6e245290384270bdbe441e4431776ea716350e","subject":"Fix some more race conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9c6e245290384270bdbe441e4431776ea716350e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:32:52.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbe578f03f09a6d7f43de3d33e238f3934bb2886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbe578f03f09a6d7f43de3d33e238f3934bb2886"}]},"branch":"refs/heads/master"},"1e8695b5996a1caa65714e656d7493a0eee13a11":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-10-05 14:31:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/7"}}},"commit":{"parents":[{"commit":"7fdfee6bd68c48d3bfcef243bce214fc9a69e398","subject":"Fix some more race conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fdfee6bd68c48d3bfcef243bce214fc9a69e398"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 14:15:36.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e8695b5996a1caa65714e656d7493a0eee13a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e8695b5996a1caa65714e656d7493a0eee13a11"}]},"branch":"refs/heads/master"},"e445fdf2b02b267871adb8ac459971aec8496f3d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-10-05 18:10:33.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/8"}}},"commit":{"parents":[{"commit":"eabe4b46d191ce39ff986c8ae928fdecfada3b2c","subject":"Fix some more race conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eabe4b46d191ce39ff986c8ae928fdecfada3b2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 18:10:26.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e445fdf2b02b267871adb8ac459971aec8496f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e445fdf2b02b267871adb8ac459971aec8496f3d"}]},"branch":"refs/heads/master"},"164087a4935a253667053ca3d276aff7a7cc2240":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-06 09:22:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/9"}}},"commit":{"parents":[{"commit":"dca469180460c7656d08c851ba45029e693791e8","subject":"Fix some more race conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dca469180460c7656d08c851ba45029e693791e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 08:59:29.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/164087a4935a253667053ca3d276aff7a7cc2240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/164087a4935a253667053ca3d276aff7a7cc2240"}]},"branch":"refs/heads/master"},"ffe7db85fff8d6d1812a2dcbf8e9b6a8c6f7b3ae":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/930799/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/930799/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/930799/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/930799/10"}}},"commit":{"parents":[{"commit":"7fbabf998900d6d90af008b30ff752e289fd48ce","subject":"Fix some more race conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fbabf998900d6d90af008b30ff752e289fd48ce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:26:29.000000000","tz":540},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia9dc54a929ea2f69b871d099dd827d06efb22078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ffe7db85fff8d6d1812a2dcbf8e9b6a8c6f7b3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ffe7db85fff8d6d1812a2dcbf8e9b6a8c6f7b3ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
