)]}'
{"id":"openstack%2Fmistral~930956","triplet_id":"openstack%2Fmistral~master~I41e4663ba2c829638670c20c92932a56015f96e3","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 13:33:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I41e4663ba2c829638670c20c92932a56015f96e3","subject":"Fix some more race conditions","status":"MERGED","created":"2024-09-30 22:23:56.000000000","updated":"2024-10-06 14:22:37.000000000","submitted":"2024-10-06 14:22:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930956","meta_rev_id":"f705b892556b4f5f11c710bd8469434cb47c8b13","_number":930956,"virtual_id_number":930956,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 14:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 00:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1d9dcc6d5bfa991b71cf9eeb442d5869ad82a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-30 22:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b75db7c82c52ea6ea2a4d7a3cc9d9db290e7bd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 00:32:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef980de7b5484ce8bc14b7ddcefd0524\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429a4b84440742e4bc38c86f0a2f6a5a : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdf85718c6ea4cfaa7fdf0de599953de : SUCCESS in 12m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b41fe8874d70436d86fdd7a8fe8810ca : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00816b4088264ce198eab6a247c01ee0 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f6fcb5852f74e9695c4eb9b46b16fc9 : SUCCESS in 16m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/418949b3457e41b0ac11a1585fc94ba2 : SUCCESS in 35m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8124ff0c87534a6d927d5f3569909c2f : SUCCESS in 36m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55b903e724bd4d65b1fe868fa6b62fc2 : SUCCESS in 38m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1bffc81e64414ad5a40b71e358d01d10 : SUCCESS in 35m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c138a0da042346af95c3064b0ff3535b : SUCCESS in 30m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/15097c235cdd4c6fafcc2c4a097c19e7 : FAILURE in 55m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d8e281371b634ea2bfb3e4d370455f7a : SUCCESS in 40m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e9a6cafa3d4544659454121cbffadeb7 : TIMED_OUT in 2h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d1ca789a2cd79920a3ebe199d0a0e672afae2b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-01 11:22:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7e1b942ac8151833b5dd58a8b6c5c77be9bf305b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:21:23.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2adab2323f3e4fd1b8e73daccc2f87ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f831e1ba06874b0a848a8cdbe134878e : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5497daa7d3b4489b832b8be6b0749c2 : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e284f2bada14c52a3ef3427a04cfe5a : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b1c39675d140d8abd845da95172269 : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df370185bb194ddabea93d6ab96b4928 : SUCCESS in 15m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a847c5eeebd54c9f8bb12c7d52a4e0ee : SUCCESS in 42m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a9132c65ffc467ea1e941d08895212d : SUCCESS in 28m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9973690ab73b4432b05e0554eb01a46d : SUCCESS in 29m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/644185949b49430581a4b7f45b67fa5e : SUCCESS in 43m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e0af7768af8453e8c4afec1def4379c : SUCCESS in 22m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f09ea9f6dc2440cda86be3a7dbbff159 : FAILURE in 42m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8d78de3fabf642f5a3e92122e7237347 : FAILURE in 52m 17s","accounts_in_message":[],"_revision_number":2},{"id":"eb098c908f13746b2327befa53fffdf9675f0eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-01 16:07:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"72077a00f13f4d7dffb869cc3e61c9710edf54f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:15:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ed577bde9344d55870d292d0f113d8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae25c6dd2134975ad2f13342e104159 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22216ee46c71486bb59254f204db2f5d : SUCCESS in 14m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0423ab64eb414e9abc4f127a744c0443 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3c5afbfbfb4623b1041c7e1b680ed3 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9e49d5a6f1b45a1bca2f7c913d00456 : SUCCESS in 18m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e377f3d68b5240339f2f1adc776db463 : SUCCESS in 38m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c438fb10a4df45d99d53d92b51b8691b : SUCCESS in 35m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c344fbd1804f7f899b51baf2c0e0f0 : SUCCESS in 38m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/05f7bce860b3424f917402674fd804f5 : SUCCESS in 35m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7256f1007dcb4a1b8886021ce2a720b0 : SUCCESS in 50m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c11d09298cd640fb9d3ac5f8050545be : TIMED_OUT in 1h 00m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2ee17f7540eb4bb58edd0c2acb1373ae : TIMED_OUT in 1h 00m 48s","accounts_in_message":[],"_revision_number":3},{"id":"120a2a6a76b9e54d972daa6b06d139d9b7e644d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-02 21:19:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3c2571e468246e3c5e7f631864dd27bd9e751a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:09:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4a1a084f5f24e5dafc4af2d3e379af3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7770f2ad674b49beb747a43169cf0df1 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72345f0b0e0d40fb94364f31100aa9ce : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e95c5296f5a048f89d2a7d8faa8c589a : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19bfc3a546514c9e9ec7b60ef6d338b0 : FAILURE in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc25969ddcbb474cb551b6bbbc1e98c4 : SUCCESS in 13m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/228850bd4f0147cd847374c6fcb1f566 : SUCCESS in 33m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01ffa5cbca4e455e91930158e93e315f : SUCCESS in 40m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eda286420054982aeeaa64a1f7da6a3 : SUCCESS in 39m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/da5200936d134180bed4d61383746760 : SUCCESS in 36m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/85fa4de3738f4d36a685e0c20af6a065 : SUCCESS in 33m 54s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/18c6cf5848a34115b9a6465fa4cb9c63 : SUCCESS in 12m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/15765f60d9db482c951fe546ffbd2658 : FAILURE in 11m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d169bbaff8443e7ae4125d96f27c7ae : FAILURE in 11m 17s","accounts_in_message":[],"_revision_number":4},{"id":"1e1884165b1b85efc70db63e46a1d44f92abe162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-04 21:37:18.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3b776ae5fa6fd937eb8174798566714214f2bc6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:22:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2c165be81e9446ab84bd528ab35d629\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d63fc3b61c4758bbd668b1352aedaf : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b72e647a701495ca8c726d11cc8464e : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d746edf6999435d831c4648c91bdd37 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb107effa2b4e2ca8e01d30759c42e2 : FAILURE in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0bca26c84884252a9e71b92ef1ec8e8 : SUCCESS in 17m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8f69a60d85974c2d8f97e12a334c9b7b : SUCCESS in 36m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a077dce7be44490c8b195605d714eb1d : SUCCESS in 33m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/534f7e3c0a1c418191c762a8d982c61f : SUCCESS in 35m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d6a973186a074ba994d2be92fe52f910 : SUCCESS in 36m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/16c334d4b220414db8b0a5ede621e619 : SUCCESS in 35m 09s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/408b85d93ac347a18d853de7c2a8e8ba : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":5},{"id":"ffabe9582914ade85d8e872371ce25260db871ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 14:31:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ea5e7f4c3a5714964a064959e4f8c76957aebcd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:25:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/248f2d5e52004a14936515fffaed6948\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a4cc8cb8977451187d233cdf87e3de7 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba6a3bb4d2ad41f7826386a6573e9c09 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d44975564c34852af86ddd4cbdb7db5 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08aeeb7b2ef54b80bfc92728666f784a : FAILURE in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4be969d09d6640d1a63d31d3c6fd4437 : FAILURE in 14m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c1c87a8104154b0e8b96b3e4ff096b89 : SUCCESS in 33m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e60f9381cac6460c8c3d93e1cdbd26a1 : SUCCESS in 33m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b45ab687ab1d4f7a95161954276e38a2 : SUCCESS in 37m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1e0be2edb28541cea1076242d7768843 : SUCCESS in 22m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4f48dd8ec9b94514bde40c7f10b19e86 : SUCCESS in 39m 30s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/86b5161d780b417aa24f80f2367f883b : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":6},{"id":"6f52bcbc751b9e0d4636ed5b6807b6d76eb94272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 18:10:33.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a9dd5a7ade182b1d88df79275c25bdbfa131d821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 19:07:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2ce45ad2e345108f84146525d06974\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396eb2c66102427ba02853a0137d60b2 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fef3037cb5e64bfd899d6d78501cf4da : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fa8542041804c81a611e0c9bdafa65e : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9276d2e664384998905bec14a3c38885 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/538f8c3455a546849591695580cc3fa3 : SUCCESS in 12m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a8cd00d77ea646b6b19d466774d5689e : SUCCESS in 46m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/427097774bc94275b20b0f210cbd7d2d : SUCCESS in 23m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dec03c9486c4257a3e39e9da4fa3c79 : SUCCESS in 25m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/574979c6b29545e28ee85173eca43928 : SUCCESS in 41m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/587c505f90bd4fc2807224cfa408d832 : SUCCESS in 22m 56s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/cd290293a7f746e88d99b45ee9c14a71 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":7},{"id":"34168d7ea85cacad652359c8f7dde42e2c0adb58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 09:22:28.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8ada82ce4045f97a699f4c1da818ff2db6da12bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 10:10:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae80b6178ae642bbb6a36ea9af286dd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfbc0d8bc36b427fa16b51898cc012c4 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/227d38cb863d44309c883a2b76ae0eb3 : SUCCESS in 13m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78b428afea7348e6a79621887479d56f : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96dfdd0337d466990fc36541b3c0a00 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce477a955da04a5194500a99f6145343 : SUCCESS in 13m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/52dc50a1b2a341f28e8583b5c4a7381c : SUCCESS in 23m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5106e6926a5429f9ee7e62e1bf8862e : SUCCESS in 40m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0002e93e4134d779f1d4e3dc34ab4c4 : SUCCESS in 39m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/25723d29ef444973a026cbde6e336371 : SUCCESS in 35m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ed13a88ca9774f569c698b25fb4f3191 : SUCCESS in 38m 49s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/4e5ddb045d094bfd8c72e807b677720e : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":8},{"id":"4c92a1536442220571eb32a51452322e8fa4fe41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"e1c68a1735c00c368622bb56f297e254b47cddd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:26:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/886ddc2f61354257aaf3d787478d786c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36fb56a61b35483dae60ee3127b3bbdd : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab156eb2fb1e4028a6a89cd812d1d847 : SUCCESS in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0de012ad95a94518bc25648f7e741fa5 : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f6e6519053459fadd16d26d6745c88 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4503aadf8b84c90848ca8bd831fded7 : FAILURE in 14m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ebf9a495a0b5422888f40208aa9eddee : SUCCESS in 24m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a5c159ed86c4df0afb1f2e9b5741ee7 : SUCCESS in 29m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73dcf3d384154403b6d0e7a0a6cd1a0e : SUCCESS in 26m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6c7ada3147554cb28bfdecd15a0229f9 : SUCCESS in 22m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7fe039104d76491c8c0fb0cca2351826 : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":9},{"id":"4adc9cbf9b370d04a7e9411d8a1fc89845abb763","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:36.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"9eef6154d781f7a37ebbc23da4a7da195d3fa4aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:32.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4ae4631a18a43284d310486fcd755bed338d983f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:22:37.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3355f4ff8010404d9a7b94b526a7e37d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef6bee6b0004e729718741c82805a01 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2defe2cf5fc844b5bb03ebbce0b649a8 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75f0642a79324d60a55b1a86f12acbd7 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edaa0c620a544090b4850a22302bc93a : SUCCESS in 6m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0569ac269e0f46d483d8f0a00e84aaea : SUCCESS in 27m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/958cb24ba68847c393c2c3524dc8dbb2 : SUCCESS in 29m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0023730458344edab26e491eb40a872 : SUCCESS in 35m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c4b3db66b8474f8c874b9bb893b280de : SUCCESS in 35m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d4977ee3a6c04f1ba17cddaa9031da32 : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":9},{"id":"f705b892556b4f5f11c710bd8469434cb47c8b13","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:22:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7fbabf998900d6d90af008b30ff752e289fd48ce","revisions":{"61e055095ffa9d3d78934acd1571b1e08a1ec84c":{"kind":"REWORK","_number":1,"created":"2024-09-30 22:23:56.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/1"}}},"commit":{"parents":[{"commit":"5221f9c4166d7470408494eafb7e53710c4dc174","subject":"Allow adding some args to tox tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5221f9c4166d7470408494eafb7e53710c4dc174"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61e055095ffa9d3d78934acd1571b1e08a1ec84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61e055095ffa9d3d78934acd1571b1e08a1ec84c"}]},"branch":"refs/heads/master"},"3a3fdf3299bf826b7f6e36c6c2bfeaae7962823a":{"kind":"REWORK","_number":2,"created":"2024-10-01 11:22:27.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/2"}}},"commit":{"parents":[{"commit":"cd9d4a3b7eca47470facdee7091bd8a1470270b9","subject":"Remove rally test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd9d4a3b7eca47470facdee7091bd8a1470270b9"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-01 11:20:25.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3a3fdf3299bf826b7f6e36c6c2bfeaae7962823a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3a3fdf3299bf826b7f6e36c6c2bfeaae7962823a"}]},"branch":"refs/heads/master"},"462b0faef2660241ab696821751d840ac430721c":{"kind":"REWORK","_number":3,"created":"2024-10-01 16:07:24.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/3"}}},"commit":{"parents":[{"commit":"cd9d4a3b7eca47470facdee7091bd8a1470270b9","subject":"Remove rally test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd9d4a3b7eca47470facdee7091bd8a1470270b9"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-10-01 16:07:19.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/462b0faef2660241ab696821751d840ac430721c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/462b0faef2660241ab696821751d840ac430721c"}]},"branch":"refs/heads/master"},"a4e36894f6443977b35195bf5061f80db193fbed":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-02 21:19:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/4"}}},"commit":{"parents":[{"commit":"68c29e57234b4e4b7de251ee756b9cce68ad4de2","subject":"Rework tox unit tests config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68c29e57234b4e4b7de251ee756b9cce68ad4de2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 21:17:38.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a4e36894f6443977b35195bf5061f80db193fbed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a4e36894f6443977b35195bf5061f80db193fbed"}]},"branch":"refs/heads/master"},"9c6e245290384270bdbe441e4431776ea716350e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-04 21:37:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/5"}}},"commit":{"parents":[{"commit":"4587eac53d3bb013ecc9ee2c78619536834ad613","subject":"Rework tox unit tests config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4587eac53d3bb013ecc9ee2c78619536834ad613"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:32:52.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9c6e245290384270bdbe441e4431776ea716350e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9c6e245290384270bdbe441e4431776ea716350e"}]},"branch":"refs/heads/master"},"7fdfee6bd68c48d3bfcef243bce214fc9a69e398":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-05 14:31:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/6"}}},"commit":{"parents":[{"commit":"0cd5bd04103b3f72eb0e034bdbbc176a75b45203","subject":"Rework tox unit tests config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0cd5bd04103b3f72eb0e034bdbbc176a75b45203"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 14:15:36.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fdfee6bd68c48d3bfcef243bce214fc9a69e398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fdfee6bd68c48d3bfcef243bce214fc9a69e398"}]},"branch":"refs/heads/master"},"eabe4b46d191ce39ff986c8ae928fdecfada3b2c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-10-05 18:10:33.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/7"}}},"commit":{"parents":[{"commit":"691fe2b19b7cdae8dc160ec7e17f9c8245b4c17f","subject":"Rework tox unit tests config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/691fe2b19b7cdae8dc160ec7e17f9c8245b4c17f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 18:10:26.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eabe4b46d191ce39ff986c8ae928fdecfada3b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eabe4b46d191ce39ff986c8ae928fdecfada3b2c"}]},"branch":"refs/heads/master"},"dca469180460c7656d08c851ba45029e693791e8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-10-06 09:22:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/8"}}},"commit":{"parents":[{"commit":"7d17ee744be8922f97ac7afcf90837175306cda8","subject":"Rework tox unit tests config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7d17ee744be8922f97ac7afcf90837175306cda8"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 08:59:29.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dca469180460c7656d08c851ba45029e693791e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dca469180460c7656d08c851ba45029e693791e8"}]},"branch":"refs/heads/master"},"7fbabf998900d6d90af008b30ff752e289fd48ce":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/930956/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/930956/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/930956/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/930956/9"}}},"commit":{"parents":[{"commit":"858a9aa19c0c8829ee1f3acbc8bf3d916a224419","subject":"Rework tox unit tests config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/858a9aa19c0c8829ee1f3acbc8bf3d916a224419"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:22:26.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Fix some more race conditions","message":"Fix some more race conditions\n\nSome races conditions prevented the workflow / tasks to be set to\nRUNNING during testing.\nWe explicitely need to wait for the task to be running to make sure\neverything works fine.\n\nRelated-bug: #2051040\n\nChange-Id: I41e4663ba2c829638670c20c92932a56015f96e3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fbabf998900d6d90af008b30ff752e289fd48ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fbabf998900d6d90af008b30ff752e289fd48ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
