)]}'
{"id":"openstack%2Fmistral~931232","triplet_id":"openstack%2Fmistral~master~I71c206aadc4133932fb591376b020a39a78d837d","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 13:33:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I71c206aadc4133932fb591376b020a39a78d837d","subject":"Rework tox unit tests config","status":"MERGED","created":"2024-10-02 21:19:20.000000000","updated":"2024-10-06 14:22:36.000000000","submitted":"2024-10-06 14:22:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"931232","meta_rev_id":"0653614f11a030084652db9896c48e42148ce568","_number":931232,"virtual_id_number":931232,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 14:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 21:59:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6845a0b98d8d633f58e824013487038137cf292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-02 21:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d6c9e92d5d1b6ef707f34834adcc4ec9afabe49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:59:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/529b286a335f44ab8402b2238fc39544\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac41066560545b4adda0c1f001d6e4a : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5b654e3c338480085730ad004f5b498 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8539bb6689349539fa58712496e9418 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eefa392a59844039dbdc23cd64c4983 : FAILURE in 5m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a13b0381545a436d8fed82d5168fa209 : FAILURE in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1baefcd10f4e4a96870f96119469416b : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edb647ec539e4b3abe8f4eeda068eacd : FAILURE in 14m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d7cb287b52a540139b886683d5c62f97 : SUCCESS in 35m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/733875ff99d444babead70e07d54e669 : SUCCESS in 26m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a04fc607bd3f47f4bf294cc19edcf4d6 : SUCCESS in 30m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/32eb652e4ec141ed8bc353b93d31c5c2 : SUCCESS in 28m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/801f577cc80f4db59bc6461b1845e870 : SUCCESS in 32m 15s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/c49b0bafc198498190c35fe829b3256c : SUCCESS in 12m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4f0e3ba50ac24052b8f090035cc89047 : FAILURE in 11m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6cd2de7ed7f54a2b8ce5cd240bf1f19d : FAILURE in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9c8b9beafb12f3dc79281c5657f949f4698b5ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-04 21:37:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b6e6d7444611f5aced1c1f553395ee19ef9b3ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:31:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d297aad5e2d246fc904e89186655d5c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de6dcdd513d4c14b118fe42b54208b4 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50195d9816114858b87ba5fd4090fd9e : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72d1a744ab614dcc95e8905a5916df38 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d319aecf6ae24e01b472daaddb6a9313 : FAILURE in 4m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf27ef0d5b6f49949384803a05f03a89 : FAILURE in 2m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15f410fdbc414fc89e4b8a3003ae7729 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6720f2c2644e4af7ab575d1906aab423 : FAILURE in 16m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8cba7cd0eb5c4abba224ac03dcd19dbd : SUCCESS in 38m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16faac49b7bb41cdba69bd00500e6a08 : SUCCESS in 36m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4612f215a1a94f4ca6859c0267e61e55 : SUCCESS in 25m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4e3649db5d0f4ecd97146d0d9007ba5b : SUCCESS in 25m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7dd8a58619d6473080883d1b2c124ae5 : SUCCESS in 39m 24s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/3e0a5135e05d440faeab97f4e340219a : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b7a9f0303fe799e63a0909ad5cd8802c80ae1181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 14:31:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"94c1f45e9f26fb2259b2696b5d5686a88123f36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:28:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/feb4a254eeea471ead07ad672ee68a36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c6231e5d9342ea9fdb2217ab7424e9 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/208d858e258e47b3a2aba9e94c511a1f : SUCCESS in 8m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcc0c21f612340e6acc874bbe3d670d0 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad94ef0306c745948d50f6958fd17091 : FAILURE in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acb0a131f5e5453a9ad8bed9da45a74c : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f906426038804dba9016a80f561f5f8d : FAILURE in 16m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/745303367e36422c9e101058ed11ae25 : SUCCESS in 30m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56236da1d1494955912779e207276d2a : SUCCESS in 32m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cc94020fac04d05bb3df0a3ed4de751 : SUCCESS in 45m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/13cfbb5537254bd8aa7ad42dfd8e7567 : SUCCESS in 40m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/48c98783ab564605b75a104266474556 : SUCCESS in 22m 50s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/846012d8d6594d1e9b9c4e5633ffd588 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":3},{"id":"3466046a09826b4c67b4662c63d6c32c16810754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 18:10:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"043debb4794119d2a64fd9274fe48d69b53fdc85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 18:58:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb93ce88333d49d6bd885a4b9c5c2c0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a500cc6e384b728a047be23ba7cdb9 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d4c93c8ad5b4fe8a887d2ee5ef60757 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af06f6811b4b44f4bc02f44f6bd35f0a : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247bb0e5e3094605a942410ceafe2773 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d879331879c45c6870685421df3fdce : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e756533a6ef446cba0adef5c315d562 : SUCCESS in 14m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f34c27b08a2c4c9b88279f435d0f8d44 : SUCCESS in 41m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7ee23a12c0c4483a2fdebba68b7590d : SUCCESS in 27m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aafcac2eefb448d861145bad3559630 : SUCCESS in 38m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f9c85aaa3ace457e9feabc11b34fd096 : SUCCESS in 38m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bec8b9c128e14a1baf708597d1128746 : SUCCESS in 36m 45s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/647f9493dda84af4a2cc37b1f999960a : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":4},{"id":"a6462ac2984ba9a9640c8875ca899f5162cbc848","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 19:10:34.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bab7cf1f3bd5c10ba983d3c723b02393642f7e12","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 19:11:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61464d217daf65edea915a87a985e94e8c80ea23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 09:22:28.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7f35c6b02153259410b7b27c3a8f748676b07898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 10:05:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c95835c21484815ace2ee2309c4b729\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c418e70256c4c1c8cbf46f5bab03a02 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7606de0f64e34d13ac9b94be047e3d29 : SUCCESS in 14m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f9531ac1e0a4e78a348f7f4d379a790 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0689894a654748509bf637332506efdd : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a05afe84af274665a2e7a4c0313bb348 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bdc68c0070946099d62c38493d497ea : SUCCESS in 15m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0d26484373cd4755b6baed5d574c06ea : SUCCESS in 26m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a76fa9cf70f4aaf9873230ae4d97ea0 : SUCCESS in 33m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79d6b5a81c0047f1b21a2bd26eb0d57b : SUCCESS in 29m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b2ea3d9a97b24ed1ad1621bf5a96dba9 : SUCCESS in 23m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b79514b240174ef18eaa71ab80df4b60 : SUCCESS in 29m 04s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/bb47d4f84e48456ea6a3dc0c2cefb115 : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":5},{"id":"eb268fd37ee84dfbe2d01f98df4e38c3353cadcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"56da374be2ae38af79b1327951510ad764fa9677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:17:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8561ffa63afd42a584d9d4dbfc827a3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d40f6ae8bc4cb7bb5966914c739d78 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64f7ad7d37234d64b29ec42c57f2333d : SUCCESS in 11m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/074a91854c9d43e58aa45d6e372614c8 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32ca8c3e3d0426eb4a8ec9598292ebc : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0248479adb1c47a9bd662fd8de6245e0 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ec635707dae492d8f27fedbd4d2f6dd : FAILURE in 15m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/71e92ef28d4c4639a1e18c5a166b276a : SUCCESS in 42m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ea598ed065f4411b77c86f33c9ad722 : SUCCESS in 29m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/660f3ade0b1e4769ae84cd09436808f3 : SUCCESS in 39m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cee714d0eb9a40a28abcd0d1816f78e1 : SUCCESS in 23m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1263e0f0e59f4c64928d9242fe431095 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":6},{"id":"5b55b03915a15b162c05dab7f5e433baa763f19b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"59c28dd66f6de2abcccb7dd2fe282e4ce88adbaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c3b3b5012738e1d22fc86a70f86608b7de4a4760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:22:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a155ef2da7f24e8f973c28c9a1b75c81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d647008069334c17a89090a54e2196e6 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/818c4c8acb594ec995317abaa4a37e83 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4724b80b2f3b419ab19f5fc323d56773 : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2d67208e3844538e59a66a576947a2 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa706e72fbbc4bb39d20240af138b982 : SUCCESS in 4m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/854044703e3e4aa28adc5dc6a56adfdc : SUCCESS in 26m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc1f27ca291a48e5a81f4e70b1ed186c : SUCCESS in 42m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1e19a2e76c64e3e96d077043601d366 : SUCCESS in 25m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c7d50c711c5a43a39afad8b7f4ad010c : SUCCESS in 26m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/25725c00640f449caf672c7ed0476995 : SUCCESS in 40m 24s","accounts_in_message":[],"_revision_number":6},{"id":"0653614f11a030084652db9896c48e42148ce568","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:22:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"858a9aa19c0c8829ee1f3acbc8bf3d916a224419","revisions":{"68c29e57234b4e4b7de251ee756b9cce68ad4de2":{"kind":"REWORK","_number":1,"created":"2024-10-02 21:19:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/931232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/931232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/931232/1"}}},"commit":{"parents":[{"commit":"cd9d4a3b7eca47470facdee7091bd8a1470270b9","subject":"Remove rally test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd9d4a3b7eca47470facdee7091bd8a1470270b9"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:05:59.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 21:17:38.000000000","tz":120},"subject":"Rework tox unit tests config","message":"Rework tox unit tests config\n\n* Upgrade tox to a newer version (a least 3.18.0 like nova)\n* Stop using run_tests.sh (use prepare_db.sh and stestr directly)\n* Add venv and mistral.log in gitignore\n* Add unit-sqlite tests\n\nChange-Id: I71c206aadc4133932fb591376b020a39a78d837d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68c29e57234b4e4b7de251ee756b9cce68ad4de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68c29e57234b4e4b7de251ee756b9cce68ad4de2"}]},"branch":"refs/heads/master"},"4587eac53d3bb013ecc9ee2c78619536834ad613":{"kind":"REWORK","_number":2,"created":"2024-10-04 21:37:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/931232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/931232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/931232/2"}}},"commit":{"parents":[{"commit":"ee59b7e8b583a22e2b6113b04a2f44bd2b081123","subject":"Disable mysql and postgresql unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee59b7e8b583a22e2b6113b04a2f44bd2b081123"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:05:59.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:32:52.000000000","tz":120},"subject":"Rework tox unit tests config","message":"Rework tox unit tests config\n\n* Upgrade tox to a newer version (a least 3.18.0 like nova)\n* Stop using run_tests.sh (use prepare_db.sh and stestr directly)\n* Add venv and mistral.log in gitignore\n\nChange-Id: I71c206aadc4133932fb591376b020a39a78d837d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4587eac53d3bb013ecc9ee2c78619536834ad613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4587eac53d3bb013ecc9ee2c78619536834ad613"}]},"branch":"refs/heads/master"},"0cd5bd04103b3f72eb0e034bdbbc176a75b45203":{"kind":"REWORK","_number":3,"created":"2024-10-05 14:31:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/931232/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/931232/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/931232/3"}}},"commit":{"parents":[{"commit":"dc026ca348c00844410f60026fdc19a60f0fc7c2","subject":"Disable mysql and postgresql unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc026ca348c00844410f60026fdc19a60f0fc7c2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:05:59.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 14:15:36.000000000","tz":120},"subject":"Rework tox unit tests config","message":"Rework tox unit tests config\n\n* Upgrade tox to a newer version (a least 3.18.0 like nova)\n* Stop using run_tests.sh (use prepare_db.sh and stestr directly)\n* Add venv and mistral.log in gitignore\n\nChange-Id: I71c206aadc4133932fb591376b020a39a78d837d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0cd5bd04103b3f72eb0e034bdbbc176a75b45203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0cd5bd04103b3f72eb0e034bdbbc176a75b45203"}]},"branch":"refs/heads/master"},"691fe2b19b7cdae8dc160ec7e17f9c8245b4c17f":{"kind":"REWORK","_number":4,"created":"2024-10-05 18:10:33.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/931232/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/931232/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/931232/4"}}},"commit":{"parents":[{"commit":"dc026ca348c00844410f60026fdc19a60f0fc7c2","subject":"Disable mysql and postgresql unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc026ca348c00844410f60026fdc19a60f0fc7c2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:05:59.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 18:10:26.000000000","tz":120},"subject":"Rework tox unit tests config","message":"Rework tox unit tests config\n\n* Upgrade tox to a newer version (a least 3.18.0 like nova)\n* Stop using run_tests.sh (use prepare_db.sh and stestr directly)\n* Add venv and mistral.log in gitignore\n\nChange-Id: I71c206aadc4133932fb591376b020a39a78d837d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/691fe2b19b7cdae8dc160ec7e17f9c8245b4c17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/691fe2b19b7cdae8dc160ec7e17f9c8245b4c17f"}]},"branch":"refs/heads/master"},"7d17ee744be8922f97ac7afcf90837175306cda8":{"kind":"REWORK","_number":5,"created":"2024-10-06 09:22:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/931232/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/931232/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/931232/5"}}},"commit":{"parents":[{"commit":"dc026ca348c00844410f60026fdc19a60f0fc7c2","subject":"Disable mysql and postgresql unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc026ca348c00844410f60026fdc19a60f0fc7c2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:05:59.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 08:59:29.000000000","tz":120},"subject":"Rework tox unit tests config","message":"Rework tox unit tests config\n\n* Upgrade tox to a newer version (a least 3.18.0 like nova)\n* Stop using run_tests.sh (use prepare_db.sh and stestr directly)\n* Add venv and mistral.log in gitignore\n\nChange-Id: I71c206aadc4133932fb591376b020a39a78d837d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7d17ee744be8922f97ac7afcf90837175306cda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7d17ee744be8922f97ac7afcf90837175306cda8"}]},"branch":"refs/heads/master"},"858a9aa19c0c8829ee1f3acbc8bf3d916a224419":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/931232/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/931232/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/931232/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/931232/6"}}},"commit":{"parents":[{"commit":"6d96f938b50e1718540c6eb9b98e36c87117bb50","subject":"Disable mysql and postgresql unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6d96f938b50e1718540c6eb9b98e36c87117bb50"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-09-30 22:05:59.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Rework tox unit tests config","message":"Rework tox unit tests config\n\n* Upgrade tox to a newer version (a least 3.18.0 like nova)\n* Stop using run_tests.sh (use prepare_db.sh and stestr directly)\n* Add venv and mistral.log in gitignore\n\nChange-Id: I71c206aadc4133932fb591376b020a39a78d837d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/858a9aa19c0c8829ee1f3acbc8bf3d916a224419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/858a9aa19c0c8829ee1f3acbc8bf3d916a224419"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
