)]}'
{"id":"openstack%2Fmistral~931234","triplet_id":"openstack%2Fmistral~master~I8cb5afcada3002e199288bc49085611442dd78dd","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 13:33:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I8cb5afcada3002e199288bc49085611442dd78dd","subject":"Stop printing useless tests results","status":"MERGED","created":"2024-10-02 21:19:20.000000000","updated":"2024-10-06 14:24:45.000000000","submitted":"2024-10-06 14:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931234","meta_rev_id":"9206f6214dd0cf7c099801bd998af5b9088b1e2d","_number":931234,"virtual_id_number":931234,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 14:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 22:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9636b7d124e52eead19a42e7c506298b9b1f8d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-02 21:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6e09c7f37e4607f54d79a9a1ed39b35178d6d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:13:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/903b19a7fef94c468f05cfbf82694f3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3e49a738f84d14a0fcffdab09410dc : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31f873e53a884e9ab5e8aaf143d17076 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/398738068ae94a34ae30e1109b284ee6 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5d0d2135ee4478b93386c27af9875a : FAILURE in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8916bba2b20443ef9c76dc751cc03e15 : FAILURE in 13m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3b28ecd6a8394611ab0a881e5a4cf705 : SUCCESS in 39m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26156f19309b4596b6d98e9e6284b55e : SUCCESS in 40m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36625e3a1f834857bc4a92f2555fa2f6 : SUCCESS in 30m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ed01a1a5baca4143b78d359f09331385 : SUCCESS in 43m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cf9a0f27fce94861b51a2e3a9b22cbd4 : SUCCESS in 22m 55s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/547fbcd11a2b41bf8a64ebc0da70b1a1 : SUCCESS in 11m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7d4c4ca702b0442eb8ea89653c917332 : FAILURE in 11m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/06f6a3a9fe4f4e5b90d06e625202f609 : FAILURE in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"18ab5325a281ad3a35f796674ae04103750318ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-04 21:37:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"32422c7fb9f057d41ea2b91c09d7f1d4566bf4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:19:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ddf8f10c185474cbca9f75946c7fdef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/558f27ca8d364a45bb4c61c4f9c84fc4 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef3325107033402d84661c94fc502686 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4292876c7d24458ad7b0fa9ad735a76 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec9487c681a4b4689c08240f80efc96 : FAILURE in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d0350ca41cc4dffa1f356181d6f4041 : FAILURE in 16m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/55de172798cf426393654b9d8a1202c7 : SUCCESS in 31m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46554429b49d46eb86090d467fae1467 : SUCCESS in 35m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/361bf3704d884751a84672d872bf0375 : SUCCESS in 26m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cbcc0f3c7a85444c9c198afab2e8ff37 : SUCCESS in 35m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d496d0daa8d74a7c8439ff88f06249d1 : SUCCESS in 27m 17s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/b48142b09c864e6fb4a0124c59f1abed : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"571a4dc0bbab5bc9f982d3c64fa98de0fed217a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 14:31:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3d6a94b2184b46f33513f20123094a87cd0f2e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:28:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24fd02eb8dc14f04869e97bc707fe035\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a553c5e085254de6ab1a0fb0cb22da38 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/167b19b2ad884e5cb44c9c429c8fb164 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fced981375f443cb8e147db8cd72d13f : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3524d53f611748e7968687793d5eaa32 : FAILURE in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61998b31cdb943d982d92bc5daffe976 : FAILURE in 18m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/69905eed458243048e4e4112443d70e7 : SUCCESS in 35m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62c17f382ca24bb2a9518bd9c2f09522 : SUCCESS in 34m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53ce546204b5487f89b3f2922bf23879 : SUCCESS in 30m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/69d7a3fe86fc450f9ce8fcfb01cf61e2 : SUCCESS in 32m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0b9271d076aa45088f4e69320953cb98 : SUCCESS in 28m 43s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/2642a1b42dc947c89fd6dc47c55ed7a6 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4e67d81de6d7c8be12c5b2e6e50131be5ed9216e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 18:10:33.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ee5180d4ab018f33db12cbba294247a9143e951c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 19:04:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/695aa2d03ff245c5a8e80d9674fcb41e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e5640421c14e428e0cb2b6dc99d90e : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f3825ee99644439b0d8cb44eacbb14b : SUCCESS in 6m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93563398b6ae4e528126f600eea356d0 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d08e7ca0c24eda9e74bc4917394872 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e656b4c6d7f14afeaf14bff00b39b36d : SUCCESS in 16m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0ce34dfb27724dbaafb715295e1414d9 : SUCCESS in 38m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e7cae50a7e94f7eb6062175b14dd4ce : SUCCESS in 25m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2088ac0336d48e9ada31d7fdf6fa5a1 : SUCCESS in 39m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe359fe70ae040c2b44d6ac0b74933fd : SUCCESS in 24m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/faaf20f0d4c7478ea3fc971735f26762 : SUCCESS in 36m 21s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/5e751fc9778349f281ba6c58187af003 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":4},{"id":"af70beba137a1f83c81fdf9c75e1e915f380c40a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 09:22:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"85b0bfa69b8d81f7588359d275f707847a3f88d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 10:08:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/744c15eccc584eefb70a05c84cb38a04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba51860df264cc8b48d864a8a00d7b6 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f94c7674844286b6e6e8b0b4798d2c : SUCCESS in 13m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/370f91b6255442be81863757e3218782 : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd58de8c1bc4225aa666635e8a552fe : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92400fce12ef47828df12f33e91ee24d : SUCCESS in 12m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a48830b53fbb4a54b19503b69c67de71 : SUCCESS in 24m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20b7c11c311a4f65a68b16ea1b565f80 : SUCCESS in 28m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fecbbf517c494430a402269e4644618b : SUCCESS in 34m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ecc47bbab3fc43b4a2023780e5e191ae : SUCCESS in 26m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7e732e96d2a34eab9df29b63aeaf949c : SUCCESS in 23m 13s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/539d386d202f4e7993892530e2df57f7 : FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":5},{"id":"5c79b72df40f9ce9f5cbe70e051947e443bd33bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b898fbdca231884be27b801f0e851f66652b7103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:14:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5236c94eea44e6977a52f7dffb1cbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c99d37fc6bd4310b1092abc0de71158 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99db48bc0f8647f1ad1818c263c2c878 : SUCCESS in 11m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bd842ae88f54b8c8aac5c49985e9162 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae98f273e364e89a33f467f3a59ae7d : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bac414ac9df2408e8b5893fe7ce5cbd1 : SUCCESS in 16m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5b46a6d4d59941329b2542ae1ce544b5 : SUCCESS in 32m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23bc0938f72a407eb2c620f7731613dc : SUCCESS in 31m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46a86f69d8094cafa1bd320912ced0fc : SUCCESS in 28m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/50fa7a325834466aa90481cc1449ce21 : SUCCESS in 31m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/505c082c26e84a18a43095e98107967c : SUCCESS in 30m 59s","accounts_in_message":[],"_revision_number":6},{"id":"4a1adb7e8db3f7536837714ee7da5e40e05ca459","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:20.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8b97053cefaf8968fd2fe1a3bbe15f0f27ddf776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e66dc9239a8c1bb212b9284a930c441484b8810a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:53.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140b177529054992906c6ab48cf71773\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd536de0142d419db6eac557c28e15b4 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7d8ea7d9faa4f32bb4241bd97e11982 : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4e45c6212724906b05abcc8852017e5 : SUCCESS in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab861ea191054725ad9ca04cf5a2e6a9 : SUCCESS in 4m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a4a2feed7eb042d7bbd326f9c27cb5ea : SUCCESS in 25m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8790ec8d466147acaf3d01beca2f1967 : SUCCESS in 27m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27132c7fab85442faeaf4f19f725bd20 : SUCCESS in 39m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c42584e7ede4b0dae0cffa3aa566a28 : SUCCESS in 39m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6ac2328546c34ecbafd2df3d17106aa6 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":6},{"id":"2d0a1e7435cffd2186c6cfeffaa374be90147e09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"9206f6214dd0cf7c099801bd998af5b9088b1e2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:24:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43bd9c1e74b4e7d811fa78b7052b15d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935f7801ab2c46d19ef6f441ef29a169 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a6deb2c42870ddd102bb73d1b1ffcc32fecdc26d","revisions":{"9b35ef4457223c37ba02215efde5bbf787e7684c":{"kind":"REWORK","_number":1,"created":"2024-10-02 21:19:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/34/931234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/931234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/931234/1"}}},"commit":{"parents":[{"commit":"2c2b8c73042be3f558f62c76a6725dfca4c8b818","subject":"Stop printing when services starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c2b8c73042be3f558f62c76a6725dfca4c8b818"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:57:08.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 21:17:38.000000000","tz":120},"subject":"Stop printing useless tests results","message":"Stop printing useless tests results\n\nChange-Id: I8cb5afcada3002e199288bc49085611442dd78dd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b35ef4457223c37ba02215efde5bbf787e7684c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b35ef4457223c37ba02215efde5bbf787e7684c"}]},"branch":"refs/heads/master"},"da5a91c410d388566b594691ae1670c7f599cb95":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-04 21:37:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/34/931234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/931234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/931234/2"}}},"commit":{"parents":[{"commit":"4c933791ea6992edde31d35ac0ca11ce1d79576e","subject":"Stop printing when services starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c933791ea6992edde31d35ac0ca11ce1d79576e"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:57:08.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:32:52.000000000","tz":120},"subject":"Stop printing useless tests results","message":"Stop printing useless tests results\n\nChange-Id: I8cb5afcada3002e199288bc49085611442dd78dd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da5a91c410d388566b594691ae1670c7f599cb95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da5a91c410d388566b594691ae1670c7f599cb95"}]},"branch":"refs/heads/master"},"6cef2db1bd44831da6ade820735894903d4fa2d2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-05 14:31:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/34/931234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/931234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/931234/3"}}},"commit":{"parents":[{"commit":"1014932af22f4ba43699689b085b5a445c4dfc30","subject":"Stop printing when services starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1014932af22f4ba43699689b085b5a445c4dfc30"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:57:08.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 14:15:36.000000000","tz":120},"subject":"Stop printing useless tests results","message":"Stop printing useless tests results\n\nChange-Id: I8cb5afcada3002e199288bc49085611442dd78dd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6cef2db1bd44831da6ade820735894903d4fa2d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6cef2db1bd44831da6ade820735894903d4fa2d2"}]},"branch":"refs/heads/master"},"982065725aee1f7c93186c3315e182d3e35d6c6a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-05 18:10:33.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/34/931234/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/931234/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/931234/4"}}},"commit":{"parents":[{"commit":"e85aed05b47f99a3b20d72f99606ccbf1fb0b2ae","subject":"Stop printing when services starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e85aed05b47f99a3b20d72f99606ccbf1fb0b2ae"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:57:08.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 18:10:26.000000000","tz":120},"subject":"Stop printing useless tests results","message":"Stop printing useless tests results\n\nChange-Id: I8cb5afcada3002e199288bc49085611442dd78dd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/982065725aee1f7c93186c3315e182d3e35d6c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/982065725aee1f7c93186c3315e182d3e35d6c6a"}]},"branch":"refs/heads/master"},"b5069ca0b3c344edeae1280f9c78739d3ef011f9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-06 09:22:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/34/931234/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/931234/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/931234/5"}}},"commit":{"parents":[{"commit":"19e2e103b249e824f8d656bf24e4908474cbace1","subject":"Stop printing when services starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/19e2e103b249e824f8d656bf24e4908474cbace1"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:57:08.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 08:59:29.000000000","tz":120},"subject":"Stop printing useless tests results","message":"Stop printing useless tests results\n\nChange-Id: I8cb5afcada3002e199288bc49085611442dd78dd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b5069ca0b3c344edeae1280f9c78739d3ef011f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b5069ca0b3c344edeae1280f9c78739d3ef011f9"}]},"branch":"refs/heads/master"},"a6deb2c42870ddd102bb73d1b1ffcc32fecdc26d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/34/931234/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/931234/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/931234/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/931234/6"}}},"commit":{"parents":[{"commit":"7ce4349e9f14a37791d6db38e91e17e0550bdb46","subject":"Stop printing when services starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7ce4349e9f14a37791d6db38e91e17e0550bdb46"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:57:08.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Stop printing useless tests results","message":"Stop printing useless tests results\n\nChange-Id: I8cb5afcada3002e199288bc49085611442dd78dd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6deb2c42870ddd102bb73d1b1ffcc32fecdc26d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6deb2c42870ddd102bb73d1b1ffcc32fecdc26d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
