)]}'
{"id":"openstack%2Fmistral~931235","triplet_id":"openstack%2Fmistral~master~I469ff43f6e986502c48a98a664c9a4a6532b633c","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 13:33:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I469ff43f6e986502c48a98a664c9a4a6532b633c","subject":"Force mistral tests to log in mistral.log","status":"MERGED","created":"2024-10-02 21:19:20.000000000","updated":"2024-10-06 14:25:29.000000000","submitted":"2024-10-06 14:23:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931235","meta_rev_id":"34df64c416f7b96286475bd1ee7bbf67a4713b90","_number":931235,"virtual_id_number":931235,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 14:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 22:11:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a657ba9b85e19af5497065744d18d63ccfd897f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-02 21:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae0ba35819ab474739fd56b861b2e119bf976c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:11:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77bdb81dc77847b09b27cc4db8802b03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6533d6262ec2476dac7d35cf35d2e5b0 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1de724020a3a466994e660fd4850bf6f : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1d676808529482fb5be8081d3527762 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c676b852a4da4f06a43b9b3e2ee605ea : FAILURE in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/194d6ad7ba02440c9a0ccc2adae21832 : FAILURE in 14m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2ad099360b3a491a888a701e7a5528d8 : SUCCESS in 40m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7dda15d74d7c4cffa616bdf81e36e6fa : SUCCESS in 24m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12264ef2f3cb4d20818b9f7184c50c27 : SUCCESS in 38m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/96ff30e4755c4bebafc95ec0c0ba1b20 : SUCCESS in 41m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2ed6f328ef604644942295aec47fecbb : SUCCESS in 41m 23s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/7f29bf2337174ccfb999692043152ab9 : SUCCESS in 12m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1034ad2d99e34b3089e813bf2f523cc6 : FAILURE in 12m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f367c3002b9b4e848cd53ab152869a38 : FAILURE in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a0fd7bac80303ba35edfa06138053567134ffa09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-04 21:37:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9850f713df420f46f19793a92c6006c2f246c57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:19:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20e44572a808491280dc32f63ed118bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afddad730e1343d393c223900cd1d6a0 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9b0672ddfca4ac7ac8f7ef3fe157cd9 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2601850d6f8442e97de968e760854d4 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a8cb69cee643319210d9f6daadcc5e : FAILURE in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d11578a15b04c848379a9abf07109a7 : FAILURE in 13m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/62aa5d82f55a4fe2a4ea12be260ea861 : SUCCESS in 29m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d791a0054042e68eb189881a7da37b : SUCCESS in 31m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f99631bac7e49bc9832c53abf8295f3 : SUCCESS in 34m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/23b71a0c441848988fbddc776e0f1911 : SUCCESS in 24m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/48cda1e1c75c4198b275c4c21419f687 : SUCCESS in 32m 35s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/8c94c58af1714636898b66b333305bfc : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e5672015fe75de43bcc3299d4a0b5e6139f69a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 14:31:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9a9b36688806b92476a7eea6d089660f961981d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:24:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e58ce83a52242e3b42a1fac856afb3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce92a0068f74f4bb56cf3b6ef8af84b : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b4443f48727449f858f9f3266a0cd4e : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c01bab189aa4486d91de7481f4ca4b2d : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afd423113e34131b66f83764d433f85 : FAILURE in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/483762402e5744b6a3e5f70e3434d8b8 : FAILURE in 15m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3f95308ee7db4dd1a7859779a307bf07 : SUCCESS in 31m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ee0a89aba8e45a0a6e96cf7c82d686f : SUCCESS in 31m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba56e0273e8544b7b1df2a8bbe607496 : SUCCESS in 38m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/56f326bd78ec4015996ba5714a3f38d0 : SUCCESS in 30m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4cc847fe84da4728a249b8b0192bca73 : SUCCESS in 41m 33s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/adce4730aa7f4411b9ba86bf621f634e : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5498071fcf20f056fc1476f606a06fac2365dc63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 18:10:33.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"92cb0d92e07c139d2701f1868e2e9c8b24667ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 19:05:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0aee6c5f0604f1fae9ac265bf3ca215\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac968b3a5f740cba8a0310542f2e20d : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/758b13167120449db9aa3758ce64efd4 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f95b238664f6409aa31add32e92494aa : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a187916e2494c9e9c6cec248f636707 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69ee89ecd69345039f882b7addf02245 : SUCCESS in 13m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c6e2fabb3b164badaedb574d00342f47 : SUCCESS in 23m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/489e5b617ac94f20bc609bbd7c7b0a62 : SUCCESS in 40m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4739336daa534c5e9d0ce7d70d2aefd0 : SUCCESS in 24m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3de14cffa6134f3ba5f01bc411b29468 : SUCCESS in 26m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b0f271235c8a4cafbadb3c19abe3f61b : SUCCESS in 24m 16s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/64cbb2a1f01c48099171672e2b7a0d1d : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":4},{"id":"5101db6bbb0dfd5d0e7aa190e7b3fc7fe554a7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 09:22:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fb783303cad1fdb7a03c71d0213aa27f09de3b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 10:18:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/035369b57d91426fbcbb8100d74c399c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9cd8d46e334d0fa6c7fbfa1105a30a : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef36faf50d454a98a18fa26fb912ce95 : SUCCESS in 11m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c73531f70c324df09aa27ad4ae2b16ab : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccd7896b500429daf81cd3b181e06b9 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59313eaae5a741cb9b7daa7a80ac21e3 : SUCCESS in 14m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/80ff9a9f9f46466e898c075a7fdd3880 : SUCCESS in 42m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4c7c2bac6dc4a8bac04a0f6a52ec5dc : SUCCESS in 40m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74126037c2c34c1bbe20a81d952a6659 : SUCCESS in 40m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/622b8e63e29c474682ba092270eb7810 : SUCCESS in 42m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5fdca5b5dec64a63a6c824f7f343906b : SUCCESS in 39m 41s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/a35f8f29205146f1892149c53977e4f0 : FAILURE in 2m 49s","accounts_in_message":[],"_revision_number":5},{"id":"b63d6c4a922c4ce7f20dda3b2fc4c19449e16417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"57d044d12196f6ce5cff49cc394aff6868be2b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:26:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45940a33a114b7c8d57904471a91a98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba273ca0b214b95802a36b69f99b278 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/682757bade6b4f63bf90a3d0673a7541 : SUCCESS in 13m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b82bb2eed84049f4a1301e43d9547a23 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a0e3ed4bcb41148f45735d3bd03ba7 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/501dd0a93c26443c87f9a97070ceca36 : SUCCESS in 15m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/162409b64aeb42b9bc5c1722c6471a0b : SUCCESS in 41m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adbc322b74954be181529f06e2a68050 : SUCCESS in 26m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf57059fb42f4a67a6329a5443be51be : SUCCESS in 28m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6d7881e186ff4b609d6773d40c01c610 : SUCCESS in 35m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6b9d81f096904f8a976fa3d721ed34c7 : SUCCESS in 23m 45s","accounts_in_message":[],"_revision_number":6},{"id":"8b4fbbfd30d7673b291057c8f5c5b70d146ba6ba","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0e7cf3f684d11449a864ad6c1042a6439900e9b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2b593cc06b2cdd27fc61dff5e2c5ca09dfe39a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82157dc72fd4ec6805b93cf449ee87c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085cca0d59304be5bfdd5b7bd8359bdd : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f133a9d8021d43659b535569a6130b87 : SUCCESS in 14m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbd2809e5a2147f49f6ca8c6c96e98e3 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696275b16acc4cbd8904209eab0ae489 : SUCCESS in 7m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f1c577597a545cfade7c324c81b8a84 : SUCCESS in 27m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b574413a29147ce8ec0a3b84faeb2f6 : SUCCESS in 32m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a26bc831f26449083d01c686235896b : SUCCESS in 25m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/748dea540d3b43ee929686543011dc0e : SUCCESS in 30m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a1ce45619c224474b722da118631dbb9 : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":6},{"id":"c7985764380c5ffdb6e2df062e5d2726ea8227c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"34df64c416f7b96286475bd1ee7bbf67a4713b90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:25:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6e91705859b420a8cb26206aa61e4b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2648a3ca6d4c5999718547407d12d9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bcc071cf1300144414a3df7b2583c831801180c3","revisions":{"ac3c883eeb2055fb76c52351824d625809632fb9":{"kind":"REWORK","_number":1,"created":"2024-10-02 21:19:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/35/931235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/931235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/931235/1"}}},"commit":{"parents":[{"commit":"9b35ef4457223c37ba02215efde5bbf787e7684c","subject":"Stop printing useless tests results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b35ef4457223c37ba02215efde5bbf787e7684c"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:59:12.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 21:17:38.000000000","tz":120},"subject":"Force mistral tests to log in mistral.log","message":"Force mistral tests to log in mistral.log\n\nLet mistral unit tests worker log in mistral.log instead of stderr and\nstdout.\nThis way, unit tests results are much more easier to read and review.\n\nChange-Id: I469ff43f6e986502c48a98a664c9a4a6532b633c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ac3c883eeb2055fb76c52351824d625809632fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ac3c883eeb2055fb76c52351824d625809632fb9"}]},"branch":"refs/heads/master"},"2dbdf8693ddf23b9bde968bccdbf3a377c22d966":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-04 21:37:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/35/931235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/931235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/931235/2"}}},"commit":{"parents":[{"commit":"da5a91c410d388566b594691ae1670c7f599cb95","subject":"Stop printing useless tests results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da5a91c410d388566b594691ae1670c7f599cb95"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:59:12.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:32:52.000000000","tz":120},"subject":"Force mistral tests to log in mistral.log","message":"Force mistral tests to log in mistral.log\n\nLet mistral unit tests worker log in mistral.log instead of stderr and\nstdout.\nThis way, unit tests results are much more easier to read and review.\n\nChange-Id: I469ff43f6e986502c48a98a664c9a4a6532b633c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2dbdf8693ddf23b9bde968bccdbf3a377c22d966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2dbdf8693ddf23b9bde968bccdbf3a377c22d966"}]},"branch":"refs/heads/master"},"0f9c96997165cfd92b8deaa1041a712c63ea5114":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-05 14:31:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/35/931235/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/931235/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/931235/3"}}},"commit":{"parents":[{"commit":"6cef2db1bd44831da6ade820735894903d4fa2d2","subject":"Stop printing useless tests results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6cef2db1bd44831da6ade820735894903d4fa2d2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:59:12.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 14:15:36.000000000","tz":120},"subject":"Force mistral tests to log in mistral.log","message":"Force mistral tests to log in mistral.log\n\nLet mistral unit tests worker log in mistral.log instead of stderr and\nstdout.\nThis way, unit tests results are much more easier to read and review.\n\nChange-Id: I469ff43f6e986502c48a98a664c9a4a6532b633c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f9c96997165cfd92b8deaa1041a712c63ea5114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f9c96997165cfd92b8deaa1041a712c63ea5114"}]},"branch":"refs/heads/master"},"b3fee26b445403d7b0de401c1c72bf5113bcc63d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-05 18:10:33.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/35/931235/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/931235/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/931235/4"}}},"commit":{"parents":[{"commit":"982065725aee1f7c93186c3315e182d3e35d6c6a","subject":"Stop printing useless tests results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/982065725aee1f7c93186c3315e182d3e35d6c6a"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:59:12.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 18:10:26.000000000","tz":120},"subject":"Force mistral tests to log in mistral.log","message":"Force mistral tests to log in mistral.log\n\nLet mistral unit tests worker log in mistral.log instead of stderr and\nstdout.\nThis way, unit tests results are much more easier to read and review.\n\nChange-Id: I469ff43f6e986502c48a98a664c9a4a6532b633c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3fee26b445403d7b0de401c1c72bf5113bcc63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3fee26b445403d7b0de401c1c72bf5113bcc63d"}]},"branch":"refs/heads/master"},"97b5ebd610341263d378ee4d335260ade1839050":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-06 09:22:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/35/931235/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/931235/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/931235/5"}}},"commit":{"parents":[{"commit":"b5069ca0b3c344edeae1280f9c78739d3ef011f9","subject":"Stop printing useless tests results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b5069ca0b3c344edeae1280f9c78739d3ef011f9"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:59:12.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 08:59:29.000000000","tz":120},"subject":"Force mistral tests to log in mistral.log","message":"Force mistral tests to log in mistral.log\n\nLet mistral unit tests worker log in mistral.log instead of stderr and\nstdout.\nThis way, unit tests results are much more easier to read and review.\n\nChange-Id: I469ff43f6e986502c48a98a664c9a4a6532b633c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97b5ebd610341263d378ee4d335260ade1839050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97b5ebd610341263d378ee4d335260ade1839050"}]},"branch":"refs/heads/master"},"bcc071cf1300144414a3df7b2583c831801180c3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/35/931235/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/931235/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/931235/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/931235/6"}}},"commit":{"parents":[{"commit":"a6deb2c42870ddd102bb73d1b1ffcc32fecdc26d","subject":"Stop printing useless tests results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6deb2c42870ddd102bb73d1b1ffcc32fecdc26d"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-02 20:59:12.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Force mistral tests to log in mistral.log","message":"Force mistral tests to log in mistral.log\n\nLet mistral unit tests worker log in mistral.log instead of stderr and\nstdout.\nThis way, unit tests results are much more easier to read and review.\n\nChange-Id: I469ff43f6e986502c48a98a664c9a4a6532b633c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bcc071cf1300144414a3df7b2583c831801180c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bcc071cf1300144414a3df7b2583c831801180c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
