)]}'
{"id":"openstack%2Fmistral~931269","triplet_id":"openstack%2Fmistral~master~I472068637f34af072662bf6d97fe370661df814e","project":"openstack/mistral","branch":"master","topic":"replace_deprecated_datetime_utcnow","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2024-10-05 10:55:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I472068637f34af072662bf6d97fe370661df814e","subject":"Replace deprecated datetime.utcnow()","status":"MERGED","created":"2024-10-03 12:16:45.000000000","updated":"2024-10-05 10:56:40.000000000","submitted":"2024-10-05 10:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931269-replace_deprecated_datetime_utcnow","meta_rev_id":"6e909bf4932bd8966e1c3456b452e3a81988a3cf","_number":931269,"virtual_id_number":931269,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-05 10:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-05 09:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-05 09:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 14:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-05 09:56:58.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"e6a7d25894730dc7ebe21252164087c0adac8a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-03 12:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d31b839c34b3d2d692831ed4743c48c22d72f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:21:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3b7e3d010774085b52aadf77e889875\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abea554055f64fbcaca2fb7ecb31ce67 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2848fdd0e8a4c098dbb0b85ee19d74b : SUCCESS in 10m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94543cc6e24a4a15af3e2665fae44f5f : FAILURE in 15m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc48761fd95444892e4e10444aed645 : SUCCESS in 6m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef382046d9b146b8a7e492595f949810 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce15ba1ec2744bac8dc6c42d766c94de : SUCCESS in 14m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1259f5a01a9a4a3eb51a8f5f7b7444d3 : SUCCESS in 42m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbd5f94877374ae482010a5e4a080edd : SUCCESS in 40m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb4723c0220e421fbfa3446b79d1a9b4 : SUCCESS in 28m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5af6024209d34816b23233c651b0514c : SUCCESS in 44m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e292343251541e5b6796117710abfbf : SUCCESS in 26m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c108631c77b5474e94fce47524406d60 : SUCCESS in 39m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8b8285c460ba4efa90acc5634d1420dc : SUCCESS in 39m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/073fe205ace7422a93dfadc4689b8a99 : TIMED_OUT in 2h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"025bef223b19960ecb918987edf9eadf57b4a7b4","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-03 14:28:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e136e2c3ea85853873af5fe0d2b0112f940a8290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:40:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7414a7fee7954a32860ccef20ab47240\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac70702173f4eb3940fd3cbd4f90696 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97b41dc875e34cd5bb690398c9ba260c : SUCCESS in 13m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5ce6b200fdf4d9c9d2323edf0eb4cdb : FAILURE in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4221c55124934b65837c42a36353d48f : SUCCESS in 6m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d98ed4c0b204530a3a89769f65ca9df : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6070152fd3474da8918886e2bd475c33 : SUCCESS in 16m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/39728063212041a49c78d2250253394b : SUCCESS in 35m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01c7867507d84c84a12681a6f51b0785 : SUCCESS in 36m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc0f0374cbf840d6b5fa83ce1e4c2640 : SUCCESS in 35m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ae63783bc7e945c5a13b0f851ae87f10 : SUCCESS in 35m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/527641fc4af84bba83da26813347c7ac : SUCCESS in 45m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/87eb03c77ed0460f804b6f59eb58e67a : SUCCESS in 56m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3ab57bf9b95f4e8bad25dd710cd88855 : SUCCESS in 53m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f9baae8093874bb5a25dc8b6e12f1552 : TIMED_OUT in 2h 02m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e521cb5682b866cd92d3781b60c210a3948f252e","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-04 10:58:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0265aed065d02c2e7f4ea6534a7c531ef50a748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 13:06:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ec8e4e100a84c06a78c224aadad1329\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d820ccc29848ebbad446055cd2ae8b : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3730aa95fb5040458beaa703520e381f : SUCCESS in 11m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06983a132711487bb9b829d28106f2b3 : FAILURE in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03159695cf34f2ab349c62a5f13bf9c : SUCCESS in 5m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed3ab700e1804da7bb80856dddb4563b : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa8250e44169493f824ee6ec0fc2c7cf : SUCCESS in 14m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d6e7373a800c43af959aaf7098b47ab9 : SUCCESS in 44m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cf7583cbe4a401181072e7dc218169a : SUCCESS in 23m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d03fbb7c57b4ee4b078ab65ee23fc68 : SUCCESS in 37m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ff3738fd5cc44b9cb8e41cf4d8ee1d00 : SUCCESS in 28m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/66cafbf9a720417e94087f85cd83ffb4 : SUCCESS in 23m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/63c35a4af1484ea28c0440ee71a724e7 : SUCCESS in 58m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5387c4b072f644718894e37733e82acb : SUCCESS in 50m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1ae7f36c98ce4afa80a1ea26ed2d7e22 : TIMED_OUT in 2h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df8f22d4a53cb0910bee3747bbddc4320c28a95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-05 04:40:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"544623f6c7e48d8edd5e8ed75437feae1db94d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 06:43:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ae057cb0206465d8f0c3601cf2979bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496a248c8327478899b53876ff493d63 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26ebf95aacc84f17b902cc72bc8297d3 : SUCCESS in 9m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/216bf26f0a3940818b7a7e4408280e5a : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b5d50b313f445491257358e2a241e4 : SUCCESS in 5m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3bf8c2213d5f48d7873df8e5f6150e48 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd9651023caf4547b5e57fcfd4f8428b : SUCCESS in 12m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f41b50e3ec04b7c8bd5ddaa0059b77c : SUCCESS in 23m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddacde3a47af440fa20e7c03a1157432 : SUCCESS in 24m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdeb7345ef0a4d24afae9b1ec62f94cf : SUCCESS in 23m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/14879640586e40c8b66767f983a60465 : SUCCESS in 22m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5bb45fdf24dd4b2495fa71789d19747f : SUCCESS in 23m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ad49bdbfe7b747a4a87e097c0da38176 : SUCCESS in 38m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f5abfac209644d76ba4ac3b11b057fd7 : SUCCESS in 36m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8feea902c83e458fa59ea514168712a2 : TIMED_OUT in 2h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"490d1539b665cb11a8fe277ff36fa2d29bfc47fe","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 09:56:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eef5a1d866c5f336ac1895d7f9aaba8e000aa7af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 09:57:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12c7df5553b822ef6f178073157d5b64154faf8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 10:55:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b1488af86e545128344ab731aafdb5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a27d53382c4e29bd15f632843f7f39 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d87c3bae7e0f4e7898900cd54e321758 : SUCCESS in 14m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5101c9a7ac7481e9bac52b7962f2cf3 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceecfb2d67e643d18c9e4b140421fab8 : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea106159f12e4428ba0304149dd0e9fb : SUCCESS in 3m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/874cc3d0c26247b988408147ce872678 : SUCCESS in 37m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdc97c7e50d64e32a3ccc035f0364be5 : SUCCESS in 41m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84b0f8087d1549a1aa2aaab1ecac77b7 : SUCCESS in 32m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/019099832fbb4193b7c4fff54df254ae : SUCCESS in 42m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/455a03010e9a4eaba05fb85d98b51d39 : SUCCESS in 33m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b1835fa7d88d4c3e809fa3d36e06e5b0 : SUCCESS in 53m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0e55009a3734446ab95e9817951ea177 : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f72f793bb045600ff86279ef0e83ad7db41702bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 10:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6e909bf4932bd8966e1c3456b452e3a81988a3cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 10:56:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc728dc3dc824d67b6ec35fd3bd16ab3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624d626b346b40d1b1ee6415cc1a3e36 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51d358436f7abdf2ef93bae52d0eb9a2029ef19d","revisions":{"9f296f4c35b2118259a92b01aef3669a8f504682":{"kind":"REWORK","_number":1,"created":"2024-10-03 12:16:45.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/69/931269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/931269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/931269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/931269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/931269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/931269/1"}}},"commit":{"parents":[{"commit":"ff176e8dce1ea5b1256eaf3a25ef01110beb2a85","subject":"Merge \"Move release note file to correct place\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff176e8dce1ea5b1256eaf3a25ef01110beb2a85"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-03 11:51:06.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-03 11:51:06.000000000","tz":540},"subject":"Replace deprecated datetime.utcnow()","message":"Replace deprecated datetime.utcnow()\n\nThe datetime.utcnow() is deprecated in Python 3.12.\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow().\nThis bumps oslo.utils to 7.0.0.\n\nChange-Id: I472068637f34af072662bf6d97fe370661df814e\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f296f4c35b2118259a92b01aef3669a8f504682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f296f4c35b2118259a92b01aef3669a8f504682"}]},"branch":"refs/heads/master"},"51d358436f7abdf2ef93bae52d0eb9a2029ef19d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-05 04:40:16.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/69/931269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/931269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/931269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/931269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/931269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/931269/2"}}},"commit":{"parents":[{"commit":"183a63b725cc1c57dc151182ccc62f37d4af5a2a","subject":"Fix py312 random int range","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/183a63b725cc1c57dc151182ccc62f37d4af5a2a"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-03 11:51:06.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 04:40:16.000000000","tz":0},"subject":"Replace deprecated datetime.utcnow()","message":"Replace deprecated datetime.utcnow()\n\nThe datetime.utcnow() is deprecated in Python 3.12.\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow().\nThis bumps oslo.utils to 7.0.0.\n\nChange-Id: I472068637f34af072662bf6d97fe370661df814e\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51d358436f7abdf2ef93bae52d0eb9a2029ef19d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51d358436f7abdf2ef93bae52d0eb9a2029ef19d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
