)]}'
{"id":"openstack%2Fmistral~931529","triplet_id":"openstack%2Fmistral~master~I7d7c499c7176cbd75f02b2aff5d070733305c4cd","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 13:33:48.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I7d7c499c7176cbd75f02b2aff5d070733305c4cd","subject":"Disable mysql and postgresql unit tests","status":"MERGED","created":"2024-10-04 21:37:18.000000000","updated":"2024-10-06 14:23:19.000000000","submitted":"2024-10-06 14:21:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931529","meta_rev_id":"3c0704b882127b3622c9ddcfe45a745ceb2d15c8","_number":931529,"virtual_id_number":931529,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 14:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 22:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"908ef278dbdc604aa16847fe9dcdd31f707676b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-04 21:37:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddea9f6586a4ed9288af565d5e5e989d3141f785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:30:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c31ccd514c1141008a0f90d1bec09751\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018593d2c3d14b69b5ed1e33f7c5f765 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/913c87b9bf724c35987f5c03fd7a492b : SUCCESS in 11m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0de6c69503ae4d3881937ce65e9962dd : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598c704ca3c8442b94d4d3ced5702a38 : SUCCESS in 7m 24s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/71548063b54d41c2b725c137260bd62f : SUCCESS in 25m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da6b9107782649d9baf7f3207ed21f09 : SUCCESS in 39m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff14173c5b647f98a69d6a50cd5369d : SUCCESS in 40m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c04b6330342d4b56985fd536860becd2 : SUCCESS in 38m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/83a809be753347ed8a7f4b9607d476ab : SUCCESS in 24m 21s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/ad18fba0658e4b8189f4e0764a96fa4e : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9f9f37b9c90f33c4cacc23a18acad6c2b3d79be6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 11:54:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3dfc5ca164a19bf4dd40e190deee4659febdfc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 11:56:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a74880982485e1849a27543a80f984b1ef877e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 12:52:54.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/930997 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"3b18ddefa9732fad4a316dcac7820a6ebb566abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-05 14:31:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d574b88b1b6147b987bde6adb786dbe78aae194e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:36:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3ca64068c44163bd8e724fbb109f7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014fcd92a1c84964a780a570c80c0ca2 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1836a52380e24ba595764e368f1b4345 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60bf948e5d4d4670b294ffaa7e3ee369 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b547812dac43cf99e78b55d3f4556f : SUCCESS in 6m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0c7fdfc6422e4e27bba590967525bc44 : SUCCESS in 23m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8db8878d2a92449086e8972572e1ec7d : SUCCESS in 41m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25d24bffd882434685d6b7b2b257aff6 : SUCCESS in 31m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/aa5fd134a60e47be89e528b6885da2f3 : SUCCESS in 23m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/206c4566008a45238a5636477ca7e7fe : SUCCESS in 40m 42s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/bf4273d58c7f48c48f5c7a186d53b502 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"eae549807e99617b31664ee80515c01883ca1b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7ce298febdc2019545a9de922d6fb345ec38ad46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:13:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0307fafb08bf490cb39470dbf1c46096\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ff5650cfa440c1bd47c2b18fa645d7 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eb3a413bab04ef3a7b032f0c90f673b : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9562545b5914dd7bbaecd987050a392 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e3cd5f658e4ef5ba979cfecfc1ca9a : SUCCESS in 4m 16s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/95dafabd40aa4612851ac97e17968f08 : SUCCESS in 26m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6432ca6bab341c89127a06837c4de35 : SUCCESS in 40m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d37dc3b3f28244dbbf105aaca5deb888 : SUCCESS in 23m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5096682431684d03acd16fb2ae22279c : SUCCESS in 28m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a7b3e0856a7044b8952373e3225c1b69 : SUCCESS in 22m 39s","accounts_in_message":[],"_revision_number":3},{"id":"39a1bd486185966b76801ec5489d6625826f57e5","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a53d0fbc5220ed8a34050d58886b5557e2f811e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"36ec3170cc1805a03ca8db3a55554189db35e831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:21:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942a097549374c3a8d20db744662eb18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0115ab9dc04acd8551c0f0b7a06993 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dde734f61dbb4baf90af049f447e1135 : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc83c52b05b64f9597d0c38ea7c44dbf : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc6159c3ef24f81bf487038b87fbab8 : SUCCESS in 6m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/178cd2e6f87c4e4db8f2209d6914a2f0 : SUCCESS in 42m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d1ebaa50b554210adb786a741b39bf2 : SUCCESS in 33m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfc8a19a28b346ee918b070e485157fc : SUCCESS in 30m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2e315213a7f049838049f1957c36c189 : SUCCESS in 25m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1bc49a6874244027ac671eb53f726011 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b04fdcb4cfe6fb137e55c8fa1b8136019376d5c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:21:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3c0704b882127b3622c9ddcfe45a745ceb2d15c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:23:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9002a2744f4543ab898b2c5da1fb2323\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d237f77cfe477fb538665e4e84ee6a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d96f938b50e1718540c6eb9b98e36c87117bb50","revisions":{"ee59b7e8b583a22e2b6113b04a2f44bd2b081123":{"kind":"REWORK","_number":1,"created":"2024-10-04 21:37:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/29/931529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/29/931529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/29/931529/1"}}},"commit":{"parents":[{"commit":"cd9d4a3b7eca47470facdee7091bd8a1470270b9","subject":"Remove rally test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd9d4a3b7eca47470facdee7091bd8a1470270b9"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:19:32.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:32:49.000000000","tz":120},"subject":"Disable mysql and postgresql unit tests","message":"Disable mysql and postgresql unit tests\n\nThe purpose of units is to tests input/output of different functions /\nmethods of a program.\nThis is not to test the fact that it works using mysql or any other sql\nengine.\n\nSo, lets focus on units that are easy to run locally with sqlite (while\nmysql / postgresql needs specific installation)\n\nChange-Id: I7d7c499c7176cbd75f02b2aff5d070733305c4cd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee59b7e8b583a22e2b6113b04a2f44bd2b081123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee59b7e8b583a22e2b6113b04a2f44bd2b081123"}]},"branch":"refs/heads/master"},"dc026ca348c00844410f60026fdc19a60f0fc7c2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-05 14:31:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/29/931529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/29/931529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/29/931529/2"}}},"commit":{"parents":[{"commit":"2d3f0251816ed5737337c90f2aa9492c6d81b5fb","subject":"Remove rally test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d3f0251816ed5737337c90f2aa9492c6d81b5fb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:19:32.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-05 14:15:36.000000000","tz":120},"subject":"Disable mysql and postgresql unit tests","message":"Disable mysql and postgresql unit tests\n\nThe purpose of units is to tests input/output of different functions /\nmethods of a program.\nThis is not to test the fact that it works using mysql or any other sql\nengine.\n\nSo, lets focus on units that are easy to run locally with sqlite (while\nmysql / postgresql needs specific installation)\n\nChange-Id: I7d7c499c7176cbd75f02b2aff5d070733305c4cd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc026ca348c00844410f60026fdc19a60f0fc7c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc026ca348c00844410f60026fdc19a60f0fc7c2"}]},"branch":"refs/heads/master"},"6d96f938b50e1718540c6eb9b98e36c87117bb50":{"kind":"REWORK","_number":3,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/29/931529/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/29/931529/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/931529/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/29/931529/3"}}},"commit":{"parents":[{"commit":"2d3f0251816ed5737337c90f2aa9492c6d81b5fb","subject":"Remove rally test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d3f0251816ed5737337c90f2aa9492c6d81b5fb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-04 21:19:32.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Disable mysql and postgresql unit tests","message":"Disable mysql and postgresql unit tests\n\nThe purpose of units is to tests input/output of different functions /\nmethods of a program.\nThis is not to test the fact that it works using mysql or any other sql\nengine.\n\nSo, lets focus on units that are easy to run locally with sqlite (while\nmysql / postgresql needs specific installation)\n\nChange-Id: I7d7c499c7176cbd75f02b2aff5d070733305c4cd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6d96f938b50e1718540c6eb9b98e36c87117bb50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6d96f938b50e1718540c6eb9b98e36c87117bb50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
