)]}'
{"id":"openstack%2Fmistral~931580","triplet_id":"openstack%2Fmistral~master~I9069edad75befcd4995a1c23924e2856da56c216","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-10-06 15:39:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I9069edad75befcd4995a1c23924e2856da56c216","subject":"Add fast8 tox env","status":"MERGED","created":"2024-10-06 09:22:28.000000000","updated":"2024-10-06 17:00:56.000000000","submitted":"2024-10-06 16:59:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931580","meta_rev_id":"dee5abeb7c8ba9dcaa07cb6a9736eeb4f8bb81f2","_number":931580,"virtual_id_number":931580,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-06 16:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-06 13:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-06 13:33:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-06 10:13:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73951a6b0cb8bc33f3fa403b0abe92b8fbaf46d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 09:22:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3e91408c2a1faae27244ec84cf6414ed1a88e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 10:13:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33062b8f9d0a45f7a3f16f0a70582594\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dd03dc9b8b425ebe7d510b025b1478 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dab8e94ecb84a2aa5b273210d06b043 : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8eef33d73f484c2db72d6d63540bb53d : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b6dc53dd774c45871d6d41beca5c53 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e51e46446ee64cf4bae880d22f047b3f : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a90481da7c9d4bea970925e4baad21c6 : SUCCESS in 18m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/918f46dfcf4f4f91a4ea9de7ff759953 : SUCCESS in 41m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4032736017244ed3aee92cb98ac75eda : SUCCESS in 36m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64df00c25a7a4823891286b004374f08 : SUCCESS in 24m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2cfdcfe5acd345208dd5b7ca54b8d947 : SUCCESS in 25m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fa044589e4814ab68469b36ecf5432ea : SUCCESS in 26m 26s\n- mistral-tox-unit-sqlite https://zuul.opendev.org/t/openstack/build/435236fdf55b4704b0e1a0e9c1fd1e09 : FAILURE in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ce61a56bb2adf7db57b5a4757c588e622e480827","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 10:30:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f412bfe3693230f189be2c2118cb3a87768791de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:15:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/766eabaa78c545908dede900b5ecb982\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc98fdc8526a450d83cbd381b0efc4bd : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8264efb642b8487395f81352bd1e6317 : SUCCESS in 11m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/513cd25eed63486c8e9992be9543bb93 : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d6342dc33941129715715a9d71ac8a : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e71d863e0a03474f9489efe192a9e311 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e52042e81549c7a549d2a903a2291f : FAILURE in 14m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1672dbf49f5c465d887d1c679b333d1c : SUCCESS in 31m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91b7972b451f471e91dcc04a0e8ee923 : SUCCESS in 32m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a944bf25f0ad4bd580b4eada1c4c3cea : SUCCESS in 30m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/632f0beed02f418c92edf094c94b0d94 : SUCCESS in 22m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e0c0e18b4bef4563bebca6c8c1e26a05 : SUCCESS in 32m 16s","accounts_in_message":[],"_revision_number":2},{"id":"345711c6a64ac7c3be39eddc7cacc5231fa8d522","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 13:33:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"14caf92ed09e4c58c9140b175482f579088d0534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 13:34:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb6b64c18cdb8a8e2a9aba8f1b31391a9d781ba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 14:28:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2b0377e76744d9a91a748c9e06b3040\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b5d2caa40640e6a4cb66dbe2148ca7 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d2cb523563347bb8e106cfa1555fb44 : FAILURE in 13m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35b0c7ec05a0470894de140ee6fc9cc7 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5195aeb1f9504810b5f855c389771184 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7335a7e35244a9bba0c52f9e6a36b72 : SUCCESS in 3m 52s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cc79a33cd4e24141949e2e0a6a95ecc7 : SUCCESS in 24m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273803ec4ebe43cfb01a60ba50c33c64 : SUCCESS in 41m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6367a605233464b9b760dcbd7552eb7 : SUCCESS in 31m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d5da12c9b1b4451a9359b4500f6d9a45 : SUCCESS in 37m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7898011cc2f8465989522dd2079289a8 : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4cc25d599a3a30d91c4fa8d2f09168377371a8e2","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-06 15:39:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79fe35086a4e8840c4e26ff229865ad1dfcd85c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 16:06:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31894782e0484292b087e258f9d0b944\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4017488cf15414a841a96aaf31125bb : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba9b6b58305c4c2cb5fda7bf11bbbddc : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36898d5516414319a297fe759b0b3f8b : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ceb86e93c942da8a2e3e07bcbe9e35 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d973ff542f44d4a740efb7bfa1173f : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e4e96d1ecf449b7b616bf2f7be0eb73 : SUCCESS in 12m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f8c9c72c0b4447338f891dc36da56a43 : SUCCESS in 22m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/484ba7a20fcd40f6833c710604ce3bd7 : SUCCESS in 21m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16f7e8f5da084f028f0b4c6e99d90d70 : SUCCESS in 22m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/71b3deacfd5d4d4182b48810cb975fa1 : SUCCESS in 21m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f101718506fe425ba2b0bff9088eba03 : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":2},{"id":"73e1eb55769c1f09a158e91bfcb3f2813e48d0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 16:07:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91478e19102ad627b1ca6f09f269c94df506c7f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 16:59:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3948355f40624f8a871cd19bea664410\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e017506dd21436db869f98b0eae1ac2 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/952f0bc50679413da79da8e82b24c989 : SUCCESS in 12m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5601df71a78f4f318cf2f437a40a0785 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9cff4c6b355400c9be2ac7e7f4b6c39 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/597973c4e1bb4dffb4fcfafdb5de4145 : SUCCESS in 2m 50s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ad6edc4d8a4d4e1d86323cf892cf231b : SUCCESS in 22m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f50be97789ca41b08e59efc302e0b2e4 : SUCCESS in 23m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc4e1192513847e1a3f1186eb77c31e3 : SUCCESS in 44m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/afa177539e5f4bf992256580218e63c0 : SUCCESS in 24m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7c3c0ee857f84155aaefd6b9c7d5a9bf : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8669c45c3a4a93c4230ce20781190a55c367935d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 16:59:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dee5abeb7c8ba9dcaa07cb6a9736eeb4f8bb81f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 17:00:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161264d62fa14c5594b4c3e4a0126879\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc3247a66d348d082e2878df3b9c6b9 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11f6405507534e4a972be23aa151c120 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"800c3dbf93eab4411b78c8e38f56f1f0b8003233","revisions":{"5c21a7042f62f62bcf807be9622836fee5ca02dd":{"kind":"REWORK","_number":1,"created":"2024-10-06 09:22:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/80/931580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/931580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/931580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/931580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/931580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/931580/1"}}},"commit":{"parents":[{"commit":"97b5ebd610341263d378ee4d335260ade1839050","subject":"Force mistral tests to log in mistral.log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97b5ebd610341263d378ee4d335260ade1839050"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 09:20:20.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 09:22:15.000000000","tz":120},"subject":"Add fast8 tox env","message":"Add fast8 tox env\n\nThis was taken from the nova project, fast8 will allow running flake8\ntests on a subset of changes (only HEAD) which is quicker.\n\nChange-Id: I9069edad75befcd4995a1c23924e2856da56c216\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c21a7042f62f62bcf807be9622836fee5ca02dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c21a7042f62f62bcf807be9622836fee5ca02dd"}]},"branch":"refs/heads/master"},"800c3dbf93eab4411b78c8e38f56f1f0b8003233":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-06 10:30:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/80/931580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/931580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/931580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/931580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/931580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/931580/2"}}},"commit":{"parents":[{"commit":"bcc071cf1300144414a3df7b2583c831801180c3","subject":"Force mistral tests to log in mistral.log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bcc071cf1300144414a3df7b2583c831801180c3"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 09:20:20.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-10-06 10:29:06.000000000","tz":120},"subject":"Add fast8 tox env","message":"Add fast8 tox env\n\nThis was taken from the nova project, fast8 will allow running flake8\ntests on a subset of changes (only HEAD) which is quicker.\n\nChange-Id: I9069edad75befcd4995a1c23924e2856da56c216\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/800c3dbf93eab4411b78c8e38f56f1f0b8003233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/800c3dbf93eab4411b78c8e38f56f1f0b8003233"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
