)]}'
{"id":"openstack%2Fmistral~937417","triplet_id":"openstack%2Fmistral~master~I28af31bc7a9c96a621b56a68077710dc3581dcf2","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-12-17 20:17:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I28af31bc7a9c96a621b56a68077710dc3581dcf2","subject":"Use threading instead of eventlet for kombu server","status":"MERGED","created":"2024-12-09 23:42:46.000000000","updated":"2024-12-17 22:01:51.000000000","submitted":"2024-12-17 22:00:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937417-eventlet-removal","meta_rev_id":"9e0a029253a4bbb5e88266b243c1427adabf6019","_number":937417,"virtual_id_number":937417,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 22:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2024-12-10 13:43:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-12 13:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-12 13:37:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 00:40:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-10 13:43:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2024-12-10 16:15:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"c78a60478e1fcfac63ddab0eedf769118ed30c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-09 23:42:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d77ed0280ea4d2ccf64286062a7d647b2cefe531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 00:40:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653dcf2bc62f45cea472fb5ac479d0fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c2c1338f99420f9bb4d93ed54e2c2c : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9830c49e1d2f421ea752bef082deef98 : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63027779e4b347c4a67dc9c7b86b59d7 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94aed5479b22436289ac8bb212d1e895 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cae5ed388a34122b356bd92b27607f9 : SUCCESS in 14m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/edee77e8f8ce4783862bd1def5c849bc : SUCCESS in 26m 58s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8976510dc07a446a8ca28eb67c5a08ab : SUCCESS in 23m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1de7a481dea84182bbad32ddc5b97e80 : SUCCESS in 26m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5a35f51aed44659a83d7786230b8d9 : SUCCESS in 50m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9948d60b35db4134a449bd044b750a46 : SUCCESS in 48m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3e2fec9265b84563bec458fe7a5f9dc8 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8ce4a3126af9e42ae8627315f9861499d7351b79","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-10 13:43:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1766d7c496f918eb2aea8263878f39004cac33f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-12 13:37:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc12d1e51c5eb0808a6a9278e4adf3e599c753e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 13:39:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"571769cc6bdfe4e059721254254d957ceba1469c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 14:54:18.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/937414 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"0ea2db88dcc83196829bdd7c5aa25ddc698b3fff","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-16 18:05:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c76838632e1c03d38b67377c4ffead8d10e1b4c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 19:25:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922b9386c0ee4cdca23230d603e89252\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3bb5a9e97a540a2adb6d58a38098559 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff3f8216d17044c6be8dc09ef7bbad0c : SUCCESS in 14m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5734eefd17974fd8800aa8842f4830d4 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc1f06e022e46e4a93d890508570538 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6e60d9ca9824b2a834420890a38f62a : FAILURE in 15m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7d5581993c94473c8f1109eedabd38a0 : SUCCESS in 36m 37s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/d234b2d6919c422aae1dc41b67dac5f7 : SUCCESS in 43m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a297064ccfa6448da0599635e055978a : SUCCESS in 44m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe5233bd8d8a4526b3ccbb2346ad3852 : SUCCESS in 21m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9130b522a5ca4f1aa29131ae392b73a2 : SUCCESS in 26m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9bd49d28c55741d7826920d1fdbd72d9 : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4396a40fde3c85c936d36f4d50241784e3da6e0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 19:26:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34463775a8c47f5b0e3550ac5ef56987cc749295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 20:16:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/167fa3d7108f4cd9bc2f526a12943331\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fdd2438225848f6a3b26ad93e2cdb7c : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/401ec5d4a42a4df98bc27a421a88bb59 : SUCCESS in 11m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e86e317ca5941efa452af0e619a2222 : SUCCESS in 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a1b2e8216b477da17be952793f4755 : SUCCESS in 7m 38s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/efde43abdffb47f3a8022c366b239447 : SUCCESS in 42m 36s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8b08d67ebcda4ec58f4866092ea2f98a : SUCCESS in 32m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e750bf0f6b949d5807658b15f94bfb1 : SUCCESS in 37m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/552984cf13184347a0bd5a8ad1af7d61 : SUCCESS in 37m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/01a5f14359b0416580cc933c5e07904e : FAILURE in 44m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7b1c95a4b47647dc8f3b5a1035277867 : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5415113a35cfc6e848340e93c360e157d02917d0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-17 20:17:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a309398241dd11234b6660c87b0ffc89dcf53254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 20:55:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fa864a0f944e01aa716f9298ffb3ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ded53f96198497488ce1e4d98dbdde3 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b79a68073314418919cfee5a1578e14 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0041b2338f82441ebcc1cf0df70d738a : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4e814a21014aeab959db4fbead9de7 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/882da8b784a14f9db61f8fc123ad8dcc : SUCCESS in 14m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b13b7e54ef1b40f099616993b5a67529 : SUCCESS in 25m 53s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/7249d2d747004c67837df5a16518cab6 : SUCCESS in 21m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43d68d9511a24c70825e50dfe381bbaf : SUCCESS in 25m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3afb354aad3745449fdfcad401149e5f : SUCCESS in 25m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/35da359cdd9a41289401d13c0e16ac47 : SUCCESS in 35m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/448aad8b00ee4e1ca44e160774a15955 : SUCCESS in 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c1d5d0138798db0e25dc11c5dc49aff4d7ff7c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 20:56:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54c6a87c53e13e9cf8d0118d72ada1dd5689f619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:00:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e147a8fbf01548a083b8b9db6ba9f2a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725cf3c2b1a144908f9401a79ca6e1f1 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a82c2db9ff874e319b76d70bc8fd91d7 : SUCCESS in 12m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74f02866812a4ec5987c559cea4af16a : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda1c7006a0e49d88178d01ca0eb3dca : SUCCESS in 4m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b458d8823454c1390ffc01ae6e4467b : SUCCESS in 25m 31s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/29ad8c8b8e1b43f2ad04d20a02f0b312 : SUCCESS in 22m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d31e30ecce0c4cd89db129983270a7b0 : SUCCESS in 23m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fef380afe2c4e96939a587401a55162 : SUCCESS in 34m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/98f625322cfc40459e7c7c00ffd5104b : SUCCESS in 29m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d8723d4101de4173b0a74f9a0bc4edc9 : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":1},{"id":"67f40260be9b6ab667bebd15f78f4e3872c5ce1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:00:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e0a029253a4bbb5e88266b243c1427adabf6019","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:01:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce113aa816dd4385b869cc5d04c4e997\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d4f56051774a53ab1d7d383f9fbc1f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"451f90e43e8ca8600799d9b01f9b889afe11fbb3","revisions":{"451f90e43e8ca8600799d9b01f9b889afe11fbb3":{"kind":"REWORK","_number":1,"created":"2024-12-09 23:42:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/17/937417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/17/937417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/17/937417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/17/937417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/17/937417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/17/937417/1"}}},"commit":{"parents":[{"commit":"7d5a1d71aac62a5d0548e1d694fad53fe70acf5e","subject":"Use threading in OsloRPCServer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7d5a1d71aac62a5d0548e1d694fad53fe70acf5e"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-12-09 23:42:11.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-12-09 23:42:11.000000000","tz":60},"subject":"Use threading instead of eventlet for kombu server","message":"Use threading instead of eventlet for kombu server\n\nChange-Id: I28af31bc7a9c96a621b56a68077710dc3581dcf2\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/451f90e43e8ca8600799d9b01f9b889afe11fbb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/451f90e43e8ca8600799d9b01f9b889afe11fbb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
