)]}'
{"id":"openstack%2Fmistral~938422","triplet_id":"openstack%2Fmistral~master~Id49c195057fa9bbd758f66756d30b9668726786b","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-01-04 10:28:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Id49c195057fa9bbd758f66756d30b9668726786b","subject":"Stop verifying some tests tasks results","status":"MERGED","created":"2025-01-03 23:37:17.000000000","updated":"2025-01-04 11:20:11.000000000","submitted":"2025-01-04 11:19:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938422-eventlet-removal","meta_rev_id":"cc4baa6322e2e704e6b822fb653438bd92765f8b","_number":938422,"virtual_id_number":938422,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-04 11:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-04 10:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-04 10:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-04 00:03:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ceca49ad314d87cc6161b72ee416571061fe53e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-03 23:37:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d110281738ddfe1d6954413d5d34c0dda86052f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-04 00:03:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dee979f3e82412a96f74627698d8fcd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e64c520933594509a6dbeb86df0067c9 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cae822ae23854f219cdd0ed27296f9d5 : SUCCESS in 9m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8258d4dafe546e0a11a8a2456dd7b3f : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b2ecc6139f4ab28cd4ab9fc2874c94 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a9ec74bb49a4fe1ab8c5b6ec4e7e964 : SUCCESS in 13m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3599731aced6441a9498d66b99e761ab : SUCCESS in 23m 52s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/75455d31fcb84f539d1756238418811f : SUCCESS in 22m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928f79361c4047ccb7c9cd8dc9652542 : SUCCESS in 25m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e24a46196de4d6993b22feec3a884af : SUCCESS in 24m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/872150bdf26d4e5e9c32bad24eb715f1 : SUCCESS in 25m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/28733c1330d848389fc37416dcd2d5af : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c01e9d70721a8351b4b1f4302e4f2d31a98a77ec","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-04 09:20:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"470759af54199027e381cf06441d78ecdd12ff49","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-04 09:23:20.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6c0bf35f3ac142189f995961d15c6b198ef176c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-04 10:27:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b0df0a95780cca1f8a3032c6f4b6599350157bdb","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-04 10:28:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c66fa1cc071bf1d2aca2a1d676b090827640648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-04 10:53:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f0bb4b749a4e8f910f944e5439ac80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00348ba06858481985a847ed4cc5ea8c : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c45fba16415b43509fa6f68ee0502747 : SUCCESS in 10m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d0645d1afa845648c2b964e1c15d944 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb57369f1d845969e7d2758aa0463ca : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60690b8e18d54e5f9c353b4a86d7e9b4 : SUCCESS in 11m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9a11ec9c0fbf4a81bf06058b181dca27 : SUCCESS in 24m 20s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5d7cef5b36ec48399f550115df6616ce : SUCCESS in 24m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0883aa64502f4b2cb2b43a87820385de : SUCCESS in 25m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/794545c489ad4ff2934d5bd4d47817c5 : SUCCESS in 24m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/38aa6b58e89646f1a87134a355f631a3 : SUCCESS in 24m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eb9e948383714fbbbeb0c839d91bceff : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f6054f636bf8bcc653520ff31bbdba61ef6602b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-04 10:54:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62c7a2f62348751fb5d0b326c3058da55a09acb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-04 11:19:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c289c5969de845e4b296695fb85203aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725fdc19cead443d852045ee09dee39b : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4abd724fdb924c0292b0bca0870005ac : SUCCESS in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7afdbc8d8e34fecaa4630ef6452ffed : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4429d7c850b243998b80ca3bcf24fc65 : SUCCESS in 4m 25s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4d7bb1ac118d428881306a4072ea197d : SUCCESS in 24m 13s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/2e3cf00836e34a899e93da35eea9e9d0 : SUCCESS in 23m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ac0ddbcd8444d4c901482c332edbd3a : SUCCESS in 22m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a84763c00581407bb97f463104b7c7ff : SUCCESS in 23m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ff84142c6c924610baddb197f5c9016d : SUCCESS in 23m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/468b558afdf44529bfc39e65efcafdc7 : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":2},{"id":"77f8d96cd63b05228d73c724f27322f402de4d45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-04 11:19:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc4baa6322e2e704e6b822fb653438bd92765f8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-04 11:20:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad7f1eb0f916428a80e95acd7f6b8950\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21df0927c404179afba956abee860cf : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a332dd3530ef691c7bb4f75ca944311dad300c77","revisions":{"a9af9471ca72cb1b7bcddedffbdeea14441c5ed2":{"kind":"REWORK","_number":1,"created":"2025-01-03 23:37:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/22/938422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/938422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/938422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/938422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/938422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/938422/1"}}},"commit":{"parents":[{"commit":"74c6e19b7324df52ea3f1cac3cf427bd31a9c912","subject":"Add join tasks to test_tasks_function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74c6e19b7324df52ea3f1cac3cf427bd31a9c912"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-03 23:35:06.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-03 23:35:06.000000000","tz":60},"subject":"Stop verifying some tests tasks results","message":"Stop verifying some tests tasks results\n\nThe test test_rerun_join_with_branch_errors is randomly failing with new\neventlet due to the fact that we can\u0027t be sure about which task is\nexecuted before a task fail.\nRemember that when a task fail, the workflow fail and no other task can\ncontinue.\n\nChange-Id: Id49c195057fa9bbd758f66756d30b9668726786b\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9af9471ca72cb1b7bcddedffbdeea14441c5ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9af9471ca72cb1b7bcddedffbdeea14441c5ed2"}]},"branch":"refs/heads/master"},"a332dd3530ef691c7bb4f75ca944311dad300c77":{"kind":"REWORK","_number":2,"created":"2025-01-04 10:27:25.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/22/938422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/938422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/938422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/938422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/938422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/938422/2"}}},"commit":{"parents":[{"commit":"74c6e19b7324df52ea3f1cac3cf427bd31a9c912","subject":"Add join tasks to test_tasks_function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74c6e19b7324df52ea3f1cac3cf427bd31a9c912"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-03 23:35:06.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-04 10:27:17.000000000","tz":60},"subject":"Stop verifying some tests tasks results","message":"Stop verifying some tests tasks results\n\nThe test test_rerun_join_with_branch_errors is randomly failing with new\neventlet due to the fact that we can\u0027t be sure about which task is\nexecuted before a task fail.\nRemember that when a task fail, the workflow fail and no other task can\ncontinue.\n\nChange-Id: Id49c195057fa9bbd758f66756d30b9668726786b\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a332dd3530ef691c7bb4f75ca944311dad300c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a332dd3530ef691c7bb4f75ca944311dad300c77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
