)]}'
{"id":"openstack%2Fmistral~940580","triplet_id":"openstack%2Fmistral~master~I1d0d4a2732b930a78839b782c06326665bbf9677","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","hashtags":[],"change_id":"I1d0d4a2732b930a78839b782c06326665bbf9677","subject":"Remove eventlet from mistral launcher","status":"ABANDONED","created":"2025-02-02 11:47:16.000000000","updated":"2025-09-03 10:08:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"5d1083fa287dfdfba542f32eaaf6195dc0f5321d","_number":940580,"virtual_id_number":940580,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-03 08:38:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-02-02 12:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67746f0a83c43c8ff15e0f4ef6e3ab46235354b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-02 11:47:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19b9dd3baaeba077e1ff15bcb21ed7a3c94de4ce","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-02 11:47:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"0f7b4183a574e5decb317fc601c7446c143ac531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 12:44:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca31447753ec4f0e81a16c8d7490643f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dbe6cdd4514158afaaf146b5fedd3c : FAILURE in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2ea3e958cb04abc9abdd5915a6e6b8c : TIMED_OUT in 40m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23cd55089d93412894cda83083751cb4 : TIMED_OUT in 40m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5fb9214d5914abf889fbcf4b21b5620 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75254ab6fafb47edab50dc73dfbaee96 : TIMED_OUT in 50m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a6c2a485dde14af08250e54a8fa10bb5 : SUCCESS in 26m 24s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/d8832448a6174ecfb83cdb233ff81a70 : SUCCESS in 23m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85f70b36689d4885a8184a6e429b9364 : SUCCESS in 44m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98ee549a9f5f4ccb9706151ff4b4b1be : FAILURE in 41m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0b2531ed7bf54dec959077dadc10254d : FAILURE in 47m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8537038d2eee484f93bdeb0a3eac3a2a : SUCCESS in 43m 35s\n\nWarning:\n  Comments left for invalid file mistral/engine/post_tx_queue.py","accounts_in_message":[],"_revision_number":1},{"id":"be1e0b8e148ff5143a51e5329a475b8ddf97cb25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-02 13:24:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d77aaac83241bf33d6ee343c1da532b6552aaf0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:26:27.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2721c20e596643ac91e029819d845873\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25141c5da0040f0bebaa2fb4aecabf5 : FAILURE in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67c496853fa948ecb7c01c476ecf4515 : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6e9dd8310fb4704ae8c1703aca8f254 : TIMED_OUT in 40m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95889872ffd4ff0857cc3cd0654ea8e : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33cc01b86a68496d92f995beac04b25f : TIMED_OUT in 51m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/aee67fb578ad4eeba201b3b5c0003a13 : SUCCESS in 44m 36s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5b9847d33ff24174a45c2f0b4d809d45 : SUCCESS in 35m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51ff6d568244459684ae2d9cc2bede62 : SUCCESS in 36m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab71b4990f1145b9a9fb147d5f7cb5b6 : FAILURE in 1h 00m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/11d3a705289b480aab05cf6757c348bd : FAILURE in 44m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/46d5a9c96a0a403187cc96dfc80049dc : SUCCESS in 36m 30s\n\nWarning:\n  Comments left for invalid file mistral/engine/post_tx_queue.py","accounts_in_message":[],"_revision_number":2},{"id":"2cea493e2b99face23e02f7c499219bc0af69958","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-03 07:46:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b16bcefe343a3f49c6785af118b08ff03f4949aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 08:38:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/597c244433444ea9aa0ce3e6a4b716e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e069f354b9bd4a34a7d2db6278e63702 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d47446254e8491a9ed09639d5fbbf3b : TIMED_OUT in 40m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c8b760e45c343ae865afae56388f72a : TIMED_OUT in 40m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ce6bf0517e49af8a7117ec1576c8a2 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21288296b1604795920e0350bb3bf000 : TIMED_OUT in 50m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/50d6236a6abb44859299aaecd739699f : SUCCESS in 22m 40s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/da2bbad7ba684bc1b274584207045e70 : SUCCESS in 26m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53fec27ea104471c964c17e245eb445a : SUCCESS in 25m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fb80230726e4d759de7ab306e140028 : FAILURE in 43m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/70b8a7476266464a8d7ca0c369dd014b : FAILURE in 42m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/55401d6402514c77a6a124826a1e200f : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":3},{"id":"5d1083fa287dfdfba542f32eaaf6195dc0f5321d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 10:08:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d82ae5d549f46a140bcc1bbf0489c379430203c4","revisions":{"148ad64cc3b08be090fa053e2afba71bcbe94954":{"kind":"REWORK","_number":1,"created":"2025-02-02 11:47:16.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/80/940580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/940580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/940580/1"}}},"commit":{"parents":[{"commit":"9f4e55731be8d9e197a2e0e28ff793cc9493c611","subject":"Use threading instead of eventlet for engine queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f4e55731be8d9e197a2e0e28ff793cc9493c611"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-30 22:10:48.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-02 11:46:36.000000000","tz":60},"subject":"Remove eventlet from mistral launcher","message":"Remove eventlet from mistral launcher\n\nChange-Id: I1d0d4a2732b930a78839b782c06326665bbf9677\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/148ad64cc3b08be090fa053e2afba71bcbe94954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/148ad64cc3b08be090fa053e2afba71bcbe94954"}]},"branch":"refs/heads/master"},"add6022141648c76ad4fb00d1a4d3d7f982671ab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-02 13:24:58.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/80/940580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/940580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/940580/2"}}},"commit":{"parents":[{"commit":"652d341314841a2e804314a055456b57209651bd","subject":"Use threading instead of eventlet for engine queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/652d341314841a2e804314a055456b57209651bd"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-30 22:10:48.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-02 13:24:52.000000000","tz":60},"subject":"Remove eventlet from mistral launcher","message":"Remove eventlet from mistral launcher\n\nChange-Id: I1d0d4a2732b930a78839b782c06326665bbf9677\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/add6022141648c76ad4fb00d1a4d3d7f982671ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/add6022141648c76ad4fb00d1a4d3d7f982671ab"}]},"branch":"refs/heads/master"},"d82ae5d549f46a140bcc1bbf0489c379430203c4":{"kind":"REWORK","_number":3,"created":"2025-02-03 07:46:43.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/80/940580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/940580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/940580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/940580/3"}}},"commit":{"parents":[{"commit":"54b7b31c0b4fea97913c25f0dc0723d668464abe","subject":"Use threading instead of eventlet for engine queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54b7b31c0b4fea97913c25f0dc0723d668464abe"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-01-30 22:10:48.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-03 07:46:33.000000000","tz":60},"subject":"Remove eventlet from mistral launcher","message":"Remove eventlet from mistral launcher\n\nChange-Id: I1d0d4a2732b930a78839b782c06326665bbf9677\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d82ae5d549f46a140bcc1bbf0489c379430203c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d82ae5d549f46a140bcc1bbf0489c379430203c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
