)]}'
{"id":"openstack%2Fmistral~940581","triplet_id":"openstack%2Fmistral~master~Ic55738eae4b275798b259598b0b648c509e8cd3a","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-04 19:13:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ic55738eae4b275798b259598b0b648c509e8cd3a","subject":"Stop using eventlet for service events","status":"MERGED","created":"2025-02-02 12:01:57.000000000","updated":"2025-02-04 20:02:39.000000000","submitted":"2025-02-04 20:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940581-eventlet-removal","meta_rev_id":"f5ec2d8165eb4334f62fb83d183055af3e89d565","_number":940581,"virtual_id_number":940581,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-04 20:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-04 14:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-04 19:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-04 19:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-02 12:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-04 14:24:11.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"e51538d96e7ec8e116e5e2dc6fc6456cdf46bbba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-02 12:01:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed169a4998b460f870dcf707c706fc6d694d4371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 12:58:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91215bd9f9184319813e65e313cca7e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f6dd5b52ee4426ebad7041dbde79577 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69898ca78f304223a88090430adb29ed : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5665367a655d4e96b78659fdca5612be : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f14071a6224ddf8f732b5102751b1b : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25c490e7d194a96a1d144733be59100 : FAILURE in 13m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d23cdda0d1234ba28090a7772ebac243 : SUCCESS in 27m 38s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/deb73d793d1f417bac5d565210d102a6 : SUCCESS in 42m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05a6012097a7408a8412b2e670701fef : SUCCESS in 43m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dd3b9e87546457cb0953adcd12e61f7 : SUCCESS in 40m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fdc5def128be474ea67c2ffdaf09df4d : SUCCESS in 25m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ecaa5c4cfb13412e8263ccf50797e7b0 : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"37f4cadfe5d40479f62620337306e0d9bf2d1010","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-04 14:24:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3657b7c027f12fda32b90d5a92cc642b8940d0a8","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-04 19:13:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"519fa3e6dfbb9f0e52d6bea08f45c7e51e53ea76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 19:13:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b122b07b378ab054bd09d648eaf7bcdcad3d4ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 20:01:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4b45bc3b2e447afab79cb39c1fdf420\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8843b24623e04097a5ae0776cf4594b3 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cac696a6dff54350b0461938e7b20428 : SUCCESS in 12m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b429e4aab3ef414d9728fbc3998e7392 : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc0df10a62e4af7bb270193926f622d : SUCCESS in 5m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/85fcf931cd184cdc985eeb3d555f0e7e : SUCCESS in 30m 43s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/14cd1e55eade41e2b518d6aa2bbf0bde : SUCCESS in 40m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08aec38a28ed470baf29e3ea97fcb89e : SUCCESS in 33m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87c76fc17cd04c3cae5647baebd8578e : SUCCESS in 23m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d73b523f13f34fb0ad4185254e8ff909 : SUCCESS in 42m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/70e4c59d662d405fa310a2843927164b : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d58ae61cb84a77f05d5d345e7b06293763883d48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 20:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5ec2d8165eb4334f62fb83d183055af3e89d565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 20:02:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5bdc298d9f447fe8fe39a5490635b13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1e69fe5dc140bda4acb8d871e63c9b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1037f94e0164c1d65b17c48d97e0ba0e89b59de3","revisions":{"1037f94e0164c1d65b17c48d97e0ba0e89b59de3":{"kind":"REWORK","_number":1,"created":"2025-02-02 12:01:57.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/940581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/940581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/940581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/940581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/940581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/940581/1"}}},"commit":{"parents":[{"commit":"cc9371a32f55cfee61e65530d015cd038b7c252b","subject":"Merge \"Replace occurence of eventlet.sleep by time.sleep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cc9371a32f55cfee61e65530d015cd038b7c252b"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-02 12:00:39.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-02 12:00:39.000000000","tz":60},"subject":"Stop using eventlet for service events","message":"Stop using eventlet for service events\n\nChange-Id: Ic55738eae4b275798b259598b0b648c509e8cd3a\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1037f94e0164c1d65b17c48d97e0ba0e89b59de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1037f94e0164c1d65b17c48d97e0ba0e89b59de3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
