)]}'
{"id":"openstack%2Fmistral~941050","triplet_id":"openstack%2Fmistral~master~Ic4471345f3dd800c96e216139f1fcd2dfd7d3a86","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-11 09:55:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ic4471345f3dd800c96e216139f1fcd2dfd7d3a86","subject":"Generate sample config in an existing folder","status":"MERGED","created":"2025-02-08 17:39:15.000000000","updated":"2025-02-11 11:55:32.000000000","submitted":"2025-02-11 11:54:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941050","meta_rev_id":"5cf5dbe03c3cab5b1d25b317f6de18fc00917511","_number":941050,"virtual_id_number":941050,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 11:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-10 13:50:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 09:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 09:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-08 18:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 13:50:13.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"3d761cee0afd2e5073ad24479aaca8ff547b57b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-08 17:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32b3d2c4714fa5819486d3507f2e8ef9e36249fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 18:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b64c5517d04460fbfacfaca8ba38851\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b3a561bb6e4aa88005c831a0bb7580 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bea73c36e8a34925ac728d535042cc31 : SUCCESS in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90bc8467faf3477bb5a5aecc79f16a2a : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7f46c6afc74a12b376ca29ba633083 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/144a94485f384cc9a3ecd1592e463b84 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f15d5c7459486db0ff717faf442589 : SUCCESS in 14m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7af6c69560e6434ebe5b6a0e8e539aaa : SUCCESS in 43m 51s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/4464e4334bea442e8f2f7a4b8b1f4ab0 : SUCCESS in 40m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62ded9a1d0814ea091804a84dc57d732 : SUCCESS in 26m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab4106abdf3c4474b6165212c7836ab2 : SUCCESS in 41m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/aa2f1eed9f614cb5bd8d9ffd17a28351 : SUCCESS in 43m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/21e266aa49ad4f19b52eb7a198a3665f : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0de30597d8e2d05c9774af2e46ccb5d97e5d416f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-10 09:17:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0415f23a4ef49ed72ea2129d295cd9765a09dd6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 10:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e0e93acc21343a38263add47d20186f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505edb4189fd457a9f6499b4b2c829ed : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79b30fb41bab412884335c58791be45f : SUCCESS in 10m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2957de4854a5416fba322ed3535b46f2 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e788e7a9ff10413d80bbcd34a831e54d : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/352d09b646f54d80bd04bcdd44c11ea9 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1448c2606b0a49009709e24061ca3ce8 : FAILURE in 18m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b2634bd3646444c7b92d85c99ca7d3a5 : SUCCESS in 33m 27s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/ff7c33e65c644f73bc0ec7ed2449c174 : SUCCESS in 34m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f07243f737f4270a3e281ca6557dc72 : SUCCESS in 29m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12a18b788a35455a83745e1981c5b47c : SUCCESS in 35m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c6f04e26a68f4b2c96d9f9f75b0546db : SUCCESS in 32m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8fb6a6588b9a44fda6671f2d2286c24a : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b411b4a344c87137bd7b24796513db35bfa60c96","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-10 13:50:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f225e6526b40ff120e1330e4e8adfcb60c9541bc","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 09:55:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54590f568d302b95512a140020a123d3c715ec92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 09:56:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4536abea21eb3849d158c3f0e4c5cc40e3f23fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:54:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da9fab2d3834bfd9d6a9ee6b6cba69c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b853073822ad4c5cb6de64e8c05366ac : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/333eec9a72df4173b81faf2cd967d4dc : SUCCESS in 11m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/358c989172e84fc18870900652020ec0 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee22e093e432465bb4064e5f6837a67d : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62debb0093e64726ace44c38e2a00301 : SUCCESS in 4m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/273552853cad43728ffd635121c5cfb7 : SUCCESS in 52m 12s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/0e580381aa11481dbf23631026c0dd9a : SUCCESS in 39m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea528c37191649caa227c50fe5fe7ca5 : SUCCESS in 30m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9d2e3522d3d4d29a0fab4818477d81d : SUCCESS in 54m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b973f27c5a7b4a2295643d4889ffa2e9 : SUCCESS in 50m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/625b7b334d104527a85d911d9efa94b4 : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":2},{"id":"60a62079b07bea983dd8ef36ab14b03dedbab308","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:54:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5cf5dbe03c3cab5b1d25b317f6de18fc00917511","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:55:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b47b9851b740278464821bac358802\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dde3bbd1d58441d9c1706e57ed478b8 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/118c364100cb48df97cb6fd51f105ca7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7c7a4575fce98aca34e270a5b277c80185db9a9","revisions":{"0590c3fe8f275891a01ba22071642d5745ef73c3":{"kind":"REWORK","_number":1,"created":"2025-02-08 17:39:15.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/50/941050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/50/941050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/941050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/941050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/941050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/50/941050/1"}}},"commit":{"parents":[{"commit":"48aa547ab911fff8cbc51fa01e4f9636b31162d2","subject":"Clean install doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/48aa547ab911fff8cbc51fa01e4f9636b31162d2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 13:00:23.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 13:01:31.000000000","tz":60},"subject":"Generate sample config in an existing folder","message":"Generate sample config in an existing folder\n\nThe sample file is now generated in etc/mistral.conf.sample.\n\nChange-Id: Ic4471345f3dd800c96e216139f1fcd2dfd7d3a86\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0590c3fe8f275891a01ba22071642d5745ef73c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0590c3fe8f275891a01ba22071642d5745ef73c3"}]},"branch":"refs/heads/master"},"c7c7a4575fce98aca34e270a5b277c80185db9a9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-10 09:17:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/50/941050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/50/941050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/941050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/941050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/941050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/50/941050/2"}}},"commit":{"parents":[{"commit":"b6e80e1813c6f3f1767fcc81d7920e4c06ad8711","subject":"Merge \"Use threading instead of eventlet for engine queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6e80e1813c6f3f1767fcc81d7920e4c06ad8711"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 13:00:23.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 09:17:39.000000000","tz":60},"subject":"Generate sample config in an existing folder","message":"Generate sample config in an existing folder\n\nThe sample file is now generated in etc/mistral.conf.sample.\n\nChange-Id: Ic4471345f3dd800c96e216139f1fcd2dfd7d3a86\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c7c7a4575fce98aca34e270a5b277c80185db9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c7c7a4575fce98aca34e270a5b277c80185db9a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
