)]}'
{"id":"openstack%2Fmistral~941052","triplet_id":"openstack%2Fmistral~master~I8caeac2cf3d4d7e53121a0b9656175b98aad2692","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-11 09:55:23.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I8caeac2cf3d4d7e53121a0b9656175b98aad2692","subject":"Get rid of rpc_response_timeout_opt","status":"MERGED","created":"2025-02-08 17:39:15.000000000","updated":"2025-02-11 12:13:06.000000000","submitted":"2025-02-11 12:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941052","meta_rev_id":"b6040ac2d608f7f74f7b76b7fcdc8ae0e81baa3c","_number":941052,"virtual_id_number":941052,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 12:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-10 13:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 09:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 09:55:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-08 18:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 13:43:43.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"2ec57c5dc5e93271786c1edb0fc57ca54913ec36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-08 17:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12492f592f539e9fc7f3be5d3de0040a8fbb3470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 18:35:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f66ebe6a8314bff9004955685085a33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a2891af90be402cb564e16ebaf091ae : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a2996d573b2463cb3cf31548221bcbd : SUCCESS in 12m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9113f4e4c52b4e7e9c2bb39c2c0c1022 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2946b4135084e0e830be8c659b2117c : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f387fdb648149b58b2ed69d237f9b0b : FAILURE in 16m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1feaab4778d946caa9c5e4b23431b15b : SUCCESS in 28m 40s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/a0620849d0224c0ba2d917b739dee4a2 : SUCCESS in 32m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7182c76f2e4c4f21abafb510dd0b1a67 : SUCCESS in 39m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71325255e85e4b669ed2e5beabde1088 : SUCCESS in 44m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/37f6d7f1f82c4e78a87ad37a12244fd4 : SUCCESS in 39m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c9a9c32b9bb1463fab0a2d3e654d79ec : FAILURE in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"da2105f0dc130ea6e973c91c830f704854e55dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-09 09:50:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"db1181a80e52f8d2a3e92cb4898c77561ad279be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 10:18:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc143db159b4a3593fc5de79e315249\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb74aa71f659496886fea1ddec246b96 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2fc8cd33a9b4ae6a60fd50f129d8b99 : SUCCESS in 9m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99617c1d23aa4cc2b37e80f7c5fb277b : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4daa8c880b1469893f4ccdef16a0d5a : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1632863cf656437497e94afc3899ae2b : SUCCESS in 15m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3c5d8e53e2f84270a6d08536ccc54b83 : SUCCESS in 26m 40s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/c02555350fa7489fa8d45a3d5123ccf4 : SUCCESS in 21m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93993f01052940a59adfc4f09f315512 : SUCCESS in 25m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/995b6187c3e74a52a64596ea58097575 : SUCCESS in 22m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/73d0c5cadeaf4205a5bd99adc9e28089 : SUCCESS in 21m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5e85e01ddcc646b193ad7619a831510b : SUCCESS in 23m 04s","accounts_in_message":[],"_revision_number":2},{"id":"46e0f332e6e16d57e1f756b27ca84ef66b74cea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-10 09:17:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a43f030bcdb5448f33b5ab65c2bbc10cf65a3fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 10:11:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e56afe5b9e40a58b8767044940ee11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcccdd93b1734288a91f7a79e040f880 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5432ce9cc29c4d89b2f7a8f9db0f6b80 : SUCCESS in 12m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30f885883606450895bd05b29775fbac : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b513a311e2bd49709198df3b1ea3c7b7 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de440ffcba884db883f0040511eb9253 : SUCCESS in 14m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b06439ae247475ebba5de2dcb4681e3 : SUCCESS in 35m 24s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/4d24757e3d3c48d5b7fe1775fb0de555 : SUCCESS in 41m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e2ced9a92a04ea0ad756db392bcdf29 : SUCCESS in 46m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9761e93a51f84d7db39aa2e509b02007 : SUCCESS in 43m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dfd2194dc62e4c60b86cd74e3f38bb3f : SUCCESS in 38m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/92f781b2824443d59e887d4be070cc6e : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":3},{"id":"83c61dc1d159f5dd628eb0ab535f71ad4948a1d7","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-10 13:43:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9d7e22efc0311ce658e44ef1bea7326d5fc80aa6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 09:55:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bfa45a3b98c9314bdec84ba7f598fa16e2d6e5c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 09:56:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d4589a7648633c3762aa226a5d5f6c6bd92197e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:12:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9205db3e0f954a8aa87ad7a690a8fb01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd2cac3316f44d18e28bc742420d758 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db294e0df3a4208a6a7fc9133f5675d : SUCCESS in 13m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf1aac49ffec432295c487afb98ef428 : SUCCESS in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a952b51fb0e348d49bd755408b5720d2 : SUCCESS in 5m 55s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c5e3e2af035b43148ba8ffc8780ce1dc : SUCCESS in 57m 32s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/790d8557d4984465ae2e6a133ae25c53 : SUCCESS in 43m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/375dbcbed743475db87d278b2d644bfd : SUCCESS in 30m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49b8553e41824e01b78a7d36f719c9fd : SUCCESS in 32m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c819934a212a4dd5913d59579a44590f : SUCCESS in 31m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8d4208b04deb4ee0a3931f9f21ebfe85 : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d9e9d6885635ba6234fd7b303e09339089df4b99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:12:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b6040ac2d608f7f74f7b76b7fcdc8ae0e81baa3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:13:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10070ca572a43818fdbb8f41878d9fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0afc18dee4cc457d8a5ddd1245dac61c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"02813c3c5445b4afce87ee108dc85aef267adde2","revisions":{"5ee153b00cdc97a2ab0a5a7f83afab153f77d7db":{"kind":"REWORK","_number":1,"created":"2025-02-08 17:39:15.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/52/941052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/52/941052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/52/941052/1"}}},"commit":{"parents":[{"commit":"7a45f23713fcf2d9500c00f0f8a5a8373e313e05","subject":"Use correct number of spaces in config-reference doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a45f23713fcf2d9500c00f0f8a5a8373e313e05"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 17:33:41.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 17:34:20.000000000","tz":60},"subject":"Get rid of rpc_response_timeout_opt","message":"Get rid of rpc_response_timeout_opt\n\nThe rpc_response_timeout_opt is a duplicate of RPCClient from\noslo.messaging.\n\nChange-Id: I8caeac2cf3d4d7e53121a0b9656175b98aad2692\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ee153b00cdc97a2ab0a5a7f83afab153f77d7db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ee153b00cdc97a2ab0a5a7f83afab153f77d7db"}]},"branch":"refs/heads/master"},"f3dd326060264ab23a3fb98eabc467dd60361181":{"kind":"REWORK","_number":2,"created":"2025-02-09 09:50:45.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/52/941052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/52/941052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/52/941052/2"}}},"commit":{"parents":[{"commit":"7a45f23713fcf2d9500c00f0f8a5a8373e313e05","subject":"Use correct number of spaces in config-reference doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a45f23713fcf2d9500c00f0f8a5a8373e313e05"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 17:33:41.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-09 09:41:17.000000000","tz":60},"subject":"Get rid of rpc_response_timeout_opt","message":"Get rid of rpc_response_timeout_opt\n\nThe rpc_response_timeout_opt is a duplicate of RPCClient from\noslo.messaging.\n\nChange-Id: I8caeac2cf3d4d7e53121a0b9656175b98aad2692\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3dd326060264ab23a3fb98eabc467dd60361181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3dd326060264ab23a3fb98eabc467dd60361181"}]},"branch":"refs/heads/master"},"02813c3c5445b4afce87ee108dc85aef267adde2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-10 09:17:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/52/941052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/52/941052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/52/941052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/52/941052/3"}}},"commit":{"parents":[{"commit":"3052da16b2fe0b35820acfd76c9cba4690897560","subject":"Use correct number of spaces in config-reference doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3052da16b2fe0b35820acfd76c9cba4690897560"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 17:33:41.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 09:17:39.000000000","tz":60},"subject":"Get rid of rpc_response_timeout_opt","message":"Get rid of rpc_response_timeout_opt\n\nThe rpc_response_timeout_opt is a duplicate of RPCClient from\noslo.messaging.\n\nChange-Id: I8caeac2cf3d4d7e53121a0b9656175b98aad2692\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02813c3c5445b4afce87ee108dc85aef267adde2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02813c3c5445b4afce87ee108dc85aef267adde2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
