)]}'
{"id":"openstack%2Fmistral~941053","triplet_id":"openstack%2Fmistral~master~Iebc685d88c2de4fed3665cd0b41c0fafca6dad8a","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-11 09:55:13.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Iebc685d88c2de4fed3665cd0b41c0fafca6dad8a","subject":"Fix mistral config reference generation","status":"MERGED","created":"2025-02-08 17:39:15.000000000","updated":"2025-02-11 12:40:32.000000000","submitted":"2025-02-11 12:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941053","meta_rev_id":"5c167dd63f2628dc955ed45eb8bac735bd1671f6","_number":941053,"virtual_id_number":941053,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 12:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-10 13:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2025-02-11 08:47:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 09:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 09:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-08 18:55:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 13:46:13.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2025-02-11 08:47:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"d5ea6240ba58cf226188ec43d5e509709a6e3ded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-08 17:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f97c62bee5a3d7cdd50f00fd92abf7db5d7f37e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 18:55:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7ff7f2bcff246bd910ebbc9603fb442\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2e6056f3944f6083f590075c8e5aa5 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b21eccf5d76e499a8ec5a06e0f2c0618 : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/198759c954214063ad553b2d987e6498 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deee13ab664c40b7aea767eea750ea21 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc035a7ad084f75984ccf16bb250264 : FAILURE in 17m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2aa280042fe1491d82d9a75ec72be5b2 : SUCCESS in 26m 43s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/c3bcf8e2c5a24c3ea0d50875d412ff08 : SUCCESS in 30m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0203558be5b34f4baa9408b4c2bb5b36 : SUCCESS in 44m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0ae9eabe5d04c27b4ce5fe44ad4dc26 : SUCCESS in 36m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6a5d2d47b1f8432ba8d9c94f5a0733a9 : SUCCESS in 38m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1df11baeb91847829eabdc376547f4bf : FAILURE in 30m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0e8af2dfe7f77520b3665f35c786c223d3e1bfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-09 09:50:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2379baacf8e6cb72f16a9bc81d97acce5a26a48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 10:44:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d4c749bf8e945449744764bbef6cf15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57cc27f77bd2487c9d84fab0d528a255 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acc20202a95d4929beac95940f59bf7e : SUCCESS in 9m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70dbdb456af24202916aa5292d324883 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730be725b06b41d2a7e4c149b229cb7c : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/176af49869fe4cecac1d615d7543b9ba : SUCCESS in 12m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6bd3bb2b0b324701bf6f0469c65e8f83 : SUCCESS in 42m 13s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/f05db93c4d074df2a047161b7592a60b : SUCCESS in 23m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5ebe7b1f9fd458aa7ac26dedba4ad13 : SUCCESS in 47m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34907b0e946744cb97a8d6a64447d611 : SUCCESS in 23m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8901330304ca428e9a272b3177eaf8cb : SUCCESS in 45m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6e5275715a7f4c9f8aa0f748b10a07db : SUCCESS in 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"96a126e81b5365986b932e459c9d082549c165fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-10 09:17:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3c98f988a77d889d11d56b363efb404ebde39300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 10:20:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c40c8bc9a3a4a2994a2e68f9a9ebe46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907e75032ca44b6393164ef78fb7d3ec : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8520a9be88754434b8f57a257d626fb9 : SUCCESS in 11m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22e089a23c0a4bf793130c193b656c75 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4346be90e04b3aae8b157388f18f79 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1644d8698cdc4cf8a306c44d4c3f5f8f : FAILURE in 17m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/706cdecedd9d49fcb49387c4ec5a333c : SUCCESS in 41m 23s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/f9de1a97bc504f91b544b59769b70112 : SUCCESS in 29m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40a9baf4cd2948c2acc2750aa4f109fc : SUCCESS in 35m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3fa55fa8ac146a0a98f654be12858b3 : SUCCESS in 38m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/00e2fb84c0e143439339bdbba68f1e0a : SUCCESS in 47m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ff4a6e0b50da4175accb2efa93600f58 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":3},{"id":"98f8e7de0f6eb04acffb667811a3cfb77300288a","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-10 13:46:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ef22e4ad282ed2c32c375dd61b68944ab2384407","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-11 08:47:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8a917e779024308ff28532b6d92aa53bfa65f2fa","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 09:55:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a3c427b9199d1d90c267bab7d90a78d2fffbe7c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 09:56:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"36a317881524a602c0f67f0ae25cf7503060f669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:39:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/593b4ecb0bb24f8fb10016b4a3b94a81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a49c419e2510479d801419c123f9565c : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ca24ddcd50b4758b2b4773f664764c0 : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95111c8c0d3647059e1b387038e0ceb4 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cc9a2e2d9c4ca391ede4838c898878 : SUCCESS in 30m 17s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cab8a64caa42428db0ec50de86f4ca75 : SUCCESS in 1h 17m 35s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e46273994eed4bd39d5ebd01ab42f7c2 : SUCCESS in 44m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a42684a0f4574bcf949c59c83bc276bd : SUCCESS in 1h 17m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cb1aa54002b4fcdb21c5e6f77acecf6 : SUCCESS in 29m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/86775ba77d48404489a3786548e80934 : SUCCESS in 28m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/af7e655187be4334b8f489b87caa8974 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":3},{"id":"7189de7d650f8385aee9fc73db8ad14fb6a4224b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:39:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5c167dd63f2628dc955ed45eb8bac735bd1671f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:40:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ddd16abbe74f4eaa931ef4d2b3a6eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255866f7248945e0bb6a3b5e94e120ab : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0f2a37b3033add95b10a1861ae60a27046ca85ce","revisions":{"e8042b7d35e629fd8e7c80b21076bb4dc384fdb9":{"kind":"REWORK","_number":1,"created":"2025-02-08 17:39:15.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/53/941053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/941053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/941053/1"}}},"commit":{"parents":[{"commit":"5ee153b00cdc97a2ab0a5a7f83afab153f77d7db","subject":"Get rid of rpc_response_timeout_opt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ee153b00cdc97a2ab0a5a7f83afab153f77d7db"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 13:03:08.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 17:37:21.000000000","tz":60},"subject":"Fix mistral config reference generation","message":"Fix mistral config reference generation\n\nThe config reference was not working due to the fact that the code\nexpect an array of oslo config options, not a itertools.chain.\n\nBecause we were using an itertool.chain, the code was not continuing,\nthus preventing the mistral DEFAULT section to be file correctly.\n\nChange-Id: Iebc685d88c2de4fed3665cd0b41c0fafca6dad8a\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8042b7d35e629fd8e7c80b21076bb4dc384fdb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8042b7d35e629fd8e7c80b21076bb4dc384fdb9"}]},"branch":"refs/heads/master"},"6e0993edb83c964ed210f95a892d16eba58b62b8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-09 09:50:45.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/53/941053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/941053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/941053/2"}}},"commit":{"parents":[{"commit":"f3dd326060264ab23a3fb98eabc467dd60361181","subject":"Get rid of rpc_response_timeout_opt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3dd326060264ab23a3fb98eabc467dd60361181"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 13:03:08.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-09 09:41:17.000000000","tz":60},"subject":"Fix mistral config reference generation","message":"Fix mistral config reference generation\n\nThe config reference was not working due to the fact that the code\nexpect an array of oslo config options, not a itertools.chain.\n\nBecause we were using an itertool.chain, the code was not continuing,\nthus preventing the mistral DEFAULT section to be file correctly.\n\nChange-Id: Iebc685d88c2de4fed3665cd0b41c0fafca6dad8a\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e0993edb83c964ed210f95a892d16eba58b62b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e0993edb83c964ed210f95a892d16eba58b62b8"}]},"branch":"refs/heads/master"},"0f2a37b3033add95b10a1861ae60a27046ca85ce":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-10 09:17:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/53/941053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/941053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/941053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/941053/3"}}},"commit":{"parents":[{"commit":"02813c3c5445b4afce87ee108dc85aef267adde2","subject":"Get rid of rpc_response_timeout_opt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02813c3c5445b4afce87ee108dc85aef267adde2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-08 13:03:08.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 09:17:39.000000000","tz":60},"subject":"Fix mistral config reference generation","message":"Fix mistral config reference generation\n\nThe config reference was not working due to the fact that the code\nexpect an array of oslo config options, not a itertools.chain.\n\nBecause we were using an itertool.chain, the code was not continuing,\nthus preventing the mistral DEFAULT section to be file correctly.\n\nChange-Id: Iebc685d88c2de4fed3665cd0b41c0fafca6dad8a\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f2a37b3033add95b10a1861ae60a27046ca85ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f2a37b3033add95b10a1861ae60a27046ca85ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
