)]}'
{"id":"openstack%2Fmistral~941085","triplet_id":"openstack%2Fmistral~master~Ibf51292f3c43026a898c7ab81c9a31443bf980e6","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-11 09:54:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ibf51292f3c43026a898c7ab81c9a31443bf980e6","subject":"Get rid of _pool_cfg from kombu RPC driver","status":"MERGED","created":"2025-02-10 10:38:15.000000000","updated":"2025-02-11 12:47:09.000000000","submitted":"2025-02-11 12:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941085","meta_rev_id":"c50591df5d75dc42e4e71ef94a8ca4c3dda0be8b","_number":941085,"virtual_id_number":941085,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 12:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-10 13:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 09:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 09:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-10 11:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 13:54:52.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"43b15e383a133385cf696ba8ea51b724bd14eabc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-10 10:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de4f91ed7eada0a531f25880372e06e1ea3a0149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 11:26:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f42a6b31f87430e84d7de3f77770d99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4e77acbe3a4ce1955aa55ce5b51613 : FAILURE in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b951c4632ba349f6a59711a697a10075 : SUCCESS in 12m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f9fd86940b94c97a385b8806ecd6025 : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae28a8912294328b8880d99e1eb0e0b : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c814a1c495d14aa9bde18df7ece39aa1 : FAILURE in 16m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e10e50658e8f4b509d2eb462373d57b3 : SUCCESS in 31m 17s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e3cb7bf01aeb4137b60f403e69e5676f : SUCCESS in 35m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92234966e515443c9fd1f61c041e6972 : SUCCESS in 31m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a37d36a67c3444fa902c3a725c0b7473 : SUCCESS in 35m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d6681c5b52ee464e926b1dbcffd1cc5d : SUCCESS in 27m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0a3a7d67736540f388fc4d23ed281fed : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":1},{"id":"30b92114b05d664bc8b495c4c1bce3625c5f70f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-10 11:41:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c96a15a1851d8ede2e79b45fdc7aa6e00928a746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 12:18:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d75200b043747e9b9c9b71f3434f4c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f203ebb1725a4bfd8894367602147fde : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff20c5e181e41e6a51086078f827db3 : SUCCESS in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8526af7e710b456d93143dd91ee28a45 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654a773ad3de4a0db6d14baec0f0d581 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1772e802b43443d7be539418a31babb3 : FAILURE in 16m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/375874211c0a4659992451194a149d13 : SUCCESS in 27m 54s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/ea6b6fcf13af4f4bafa89b5ceb70c366 : SUCCESS in 33m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2228b704b5ba4dfbb293c600ca5fef74 : SUCCESS in 31m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9490a282acde4650912aef94ce6b2b28 : SUCCESS in 36m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6ed63383dacf4b888e096403b2433e39 : SUCCESS in 32m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/edb358d36dea4662a38c116173a113de : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":2},{"id":"24353d53f2ac60ccd2d3fed691704fba83f0abe0","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-10 13:54:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2facdc77c179eb82bf5ebaeb1a48005890d44175","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 09:54:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d4ec74c33acdbf74710883b26f625a64f7d4695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 09:56:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ec4a71dd15308ad8f10e00b17b1b7229592148e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:46:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a6ba522a6a401faa8f1ee1554300ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ef11f734d84460bef4818338c47d9b : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0cf5b8d6bed4a429098377daa4cab55 : SUCCESS in 12m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bbea1b1c0394e99af54bf0580659f77 : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fc693d6aa348f48c6cce8c35374c63 : SUCCESS in 6m 38s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b2c6748143f0484a8802cd6467466d0e : SUCCESS in 37m 07s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/189a754409364ba8ba99e9f76489dab1 : SUCCESS in 43m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cf8c5d9b9b541708f191463975046a6 : SUCCESS in 1h 14m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/506adad33d6641d79c4873b6232330d2 : SUCCESS in 31m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/587e58dd24934583ad20c4612f792e4f : SUCCESS in 1h 18m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6e94ab1b94ab421595e4d9f23b4d4e7e : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e8bdd28ad4f407f67601924269ef7f413702b502","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:46:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c50591df5d75dc42e4e71ef94a8ca4c3dda0be8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:47:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ac6158598b46d28e2141f7692218d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85bacfc78e346b58beb3c94bc707b63 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c27baffc1f5bbe8681795453464e4199cf3a55cb","revisions":{"2aae4deb391879441b8e4a7e3de02fec416cf5c6":{"kind":"REWORK","_number":1,"created":"2025-02-10 10:38:15.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/941085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/941085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/941085/1"}}},"commit":{"parents":[{"commit":"76d76244a4f75487301755cc29e3ae0d8df3e74c","subject":"Deprecated kombu RPC driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/76d76244a4f75487301755cc29e3ae0d8df3e74c"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 10:17:27.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 10:20:26.000000000","tz":60},"subject":"Get rid of _pool_cfg from kombu RPC driver","message":"Get rid of _pool_cfg from kombu RPC driver\n\nKombu RPC driver is overwritting some of the oslo.messaging default\nconfig parameters.\nThis is wrong, let\u0027s get rid of this until we completely remove the\nkombu RPC driver.\n\nChange-Id: Ibf51292f3c43026a898c7ab81c9a31443bf980e6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2aae4deb391879441b8e4a7e3de02fec416cf5c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2aae4deb391879441b8e4a7e3de02fec416cf5c6"}]},"branch":"refs/heads/master"},"c27baffc1f5bbe8681795453464e4199cf3a55cb":{"kind":"REWORK","_number":2,"created":"2025-02-10 11:41:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/941085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/941085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/941085/2"}}},"commit":{"parents":[{"commit":"76d76244a4f75487301755cc29e3ae0d8df3e74c","subject":"Deprecated kombu RPC driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/76d76244a4f75487301755cc29e3ae0d8df3e74c"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 10:17:27.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-10 11:41:04.000000000","tz":60},"subject":"Get rid of _pool_cfg from kombu RPC driver","message":"Get rid of _pool_cfg from kombu RPC driver\n\nKombu RPC driver is overwritting some of the oslo.messaging default\nconfig parameters.\nThis is wrong, let\u0027s get rid of this until we completely remove the\nkombu RPC driver.\n\nChange-Id: Ibf51292f3c43026a898c7ab81c9a31443bf980e6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c27baffc1f5bbe8681795453464e4199cf3a55cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c27baffc1f5bbe8681795453464e4199cf3a55cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
