)]}'
{"id":"openstack%2Fmistral~941223","triplet_id":"openstack%2Fmistral~master~I0f620ea9791c65a730feedc40a62fe83cd64cbfb","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-12 16:00:44.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I0f620ea9791c65a730feedc40a62fe83cd64cbfb","subject":"Stop using eventlet in scheduler tests","status":"MERGED","created":"2025-02-11 09:56:42.000000000","updated":"2025-02-12 17:33:00.000000000","submitted":"2025-02-12 17:32:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941223-eventlet-removal","meta_rev_id":"77d6eed972fad984a9d41dc35328c581e06b0a24","_number":941223,"virtual_id_number":941223,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 17:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-11 15:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-12 14:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-12 14:11:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 11:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 15:02:40.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"a0ad9ed5c8ddd48356eaa91759ac9705d9dd55ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 09:56:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0f0d35a484003c4e47146c11f7ad0b8cf15b61b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:51:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d82a31155504d90917b5c30fa327d45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f148ef77e83d434bb96dfe9fb15ab2eb : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ba7f9bddf284ac6ae9cd10a3fab268b : SUCCESS in 33m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1944295eb1934730be6c2b7055328b6c : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53dcc6560a224579912a5efa730943ec : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cb17ebb13154ead813ecf62104a5af7 : FAILURE in 18m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a51f3d5a94214d07b2b2e72e21bfac28 : SUCCESS in 45m 53s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/b11562fb2f904d77b23833f6150f5078 : SUCCESS in 43m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bd0db698b1044619f2e4c7899b456f4 : SUCCESS in 37m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b197b19e40374e72b848ce50a416e35a : SUCCESS in 34m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1d1b7d91b9b341fc816c605f9c5c3018 : SUCCESS in 38m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/024cafeed81645d7877d6113faa8dda0 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c2248700c368cafd8ca8b9370b49a5bf6f2d3a25","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-11 15:02:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b724c0f6bbb2dab24eb6cd8504f46b6a2422f41","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-12 14:11:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d12ecba2807b276e2d326742e4b6a933b99fc38a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:12:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81923ecf4832de27042bab7762dee7b869ecd0ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 15:15:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e948aa23e0e54b64b5bdbeb2512195fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a351282971fd4dc29298e74d50f8197e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9a1504256584877958998f47b897c23 : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52e7086ecbe64d4f94dc78758bf29cef : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a161d40d0d429bbd17d2792c400aad : SUCCESS in 8m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b5eddd2347944dbfa935c7213b940485 : FAILURE in 35m 43s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/2a91861e58654d519fe4900c07fae492 : SUCCESS in 36m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad3b8f9885504397b0c5bc1d5ea85342 : SUCCESS in 38m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9b7c087aac04dbb89b5425806d1006a : SUCCESS in 36m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4e5e1586d87c4de39de547b9b927584b : SUCCESS in 43m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a064b39ea71b4a1f9644d47eb368c137 : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9d2d417e19e039210cfb67a270449e43a42f386e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-12 16:00:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0bc1799991559ce892b827adef757aea975d9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 16:47:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17e7b6c8d6e947c3b26f86fa5b40b618\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deaf22716b644ad89c92347836609afa : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b61352f48ce4c838c7c180a9eb45918 : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27546a485d5c422fa34a7c037a08f6d4 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb47aee323945acb0297c1a0e36824c : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b4cef098af94221921622f600d75a02 : SUCCESS in 14m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/009473c37f0f49d5b4800d4cf3529b91 : SUCCESS in 41m 57s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8e053e9c5b70481d8181078d21326537 : SUCCESS in 28m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/256f01d3e9fe4f2480deebf31432bd41 : SUCCESS in 36m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e98e627ac0de46bab28af828b5e49ef5 : SUCCESS in 32m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c3b6e4c15dfa4fefbffd31036638d3f4 : SUCCESS in 46m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4445cede4e7d49df8d7b767507f14598 : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"52d72bd8d7a56d2bc7a344a76b72e0e78b278ad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 16:48:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00f0eee9aed20f02a7ed5329a626861f43bf8565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:32:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1026ffcb2eeb4cb598183fd58a244f3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ed7ff6d0914620bf2e9fe20c90268b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37f720bba318467397fd6b0d0d5d0fe2 : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0a4285b89c84d54b409f3e356b2e501 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b120f1d3bc84738a63718cebb039981 : SUCCESS in 6m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/373c44c079cf493aae278f30a53e1daa : SUCCESS in 22m 35s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/773ffdc8efc240d28ea41f9e792d6743 : SUCCESS in 25m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16c4bb7519f240519bac2be37ca90764 : SUCCESS in 24m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb7efcc921a413199f877a40f7fe3ff : SUCCESS in 30m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6fd332893fed43829353446ee7cbf998 : SUCCESS in 31m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a5dda7759703486b93aec5ea588a2660 : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"96cd31e7eb0a79c11ab0d211b691f5876fa6721b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:32:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"77d6eed972fad984a9d41dc35328c581e06b0a24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:33:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16985ddcbeb442eaf3934b3b2d00446\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4345bb93a43a498ea089273e5fc1c40b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b44114e1eb603c25a2b5116ff4378d1fdc953be7","revisions":{"b44114e1eb603c25a2b5116ff4378d1fdc953be7":{"kind":"REWORK","_number":1,"created":"2025-02-11 09:56:42.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/23/941223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/941223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/941223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/941223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/941223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/941223/1"}}},"commit":{"parents":[{"commit":"d0705da1359180acb944e93c96fa9d6bd4ef04f1","subject":"Stop preparing eventlet test thread for Kombu RPC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0705da1359180acb944e93c96fa9d6bd4ef04f1"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-11 09:53:17.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-11 09:53:17.000000000","tz":60},"subject":"Stop using eventlet in scheduler tests","message":"Stop using eventlet in scheduler tests\n\nSince we are moving away from eventlet, let\u0027s get rid of eventlet usage\nin scheduler tests.\n\nChange-Id: I0f620ea9791c65a730feedc40a62fe83cd64cbfb\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b44114e1eb603c25a2b5116ff4378d1fdc953be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b44114e1eb603c25a2b5116ff4378d1fdc953be7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
