)]}'
{"id":"openstack%2Fmistral~941232","triplet_id":"openstack%2Fmistral~master~I664c07aadb7b6fcd00c74866dd50b4662fe18206","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-12 21:56:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I664c07aadb7b6fcd00c74866dd50b4662fe18206","subject":"Use threading instead of eventlet for launcher tests","status":"MERGED","created":"2025-02-11 11:40:58.000000000","updated":"2025-02-12 23:40:14.000000000","submitted":"2025-02-12 23:39:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941232-eventlet-removal","meta_rev_id":"376a5f3d86832e6b63cd49c5f225a9dcbbf0c2eb","_number":941232,"virtual_id_number":941232,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 23:39:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-11 14:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-12 14:11:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-12 14:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 13:11:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 14:59:03.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"d6c6340c740848d3c6c4b8790aaaadccd871531f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 11:40:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d258aed1d334c1822cf7ae610ae10f2c7dc82d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 13:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7131a03563c5452a9da77278da31101f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f317783d3aff441ab13ebe5955c91e35 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f9924ac45ae4ecaa31587a384879f12 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4f0670b6dcf44a9a1e2c0eba38285f7 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96244cc8b6d44a89534fc549c3fb83a : SUCCESS in 26m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d864613850be4d50ae17ee2e7010d731 : FAILURE in 25m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/25862d4851694b6a8483731bd6b70b8e : SUCCESS in 34m 23s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/20e75af9ec5041df85e1f797c2bea600 : SUCCESS in 1h 07m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77851d2ddb3e4c2aad1755997b0a86a4 : SUCCESS in 44m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d4b5ce33354405b420d6dda2fc4b19 : SUCCESS in 41m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/496f91de89984dc68f7e7fd28e54dcd8 : SUCCESS in 40m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/647533327cab42628c0aec91fe2c1132 : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f0b81d7230eca4f487fca1703d7799953f51c777","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-11 14:59:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9fcded7ee5bf11be9ab55d9499fc43729e450ad1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-12 14:11:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a5fa3b8da00c077b742b3c7342c2fdb5485da49f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:12:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6dc5203e39727885612239c947bd5beae061262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 15:15:32.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/941223 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"9b9b6adf59b3110179a4a81dac97983845f13039","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-12 21:56:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32eca7ad6a98a1d1aab2202ccb7604fd65a23b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 22:48:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32246a22ada641ea98d70133caa8215e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab03ceb20e442a0aeb541090b2eb864 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63e6dbf881dc493dbf95a88957316785 : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/922757cb41704d2d8f09a65b0d29b494 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eeb19c169d24a4297192d56e8d86580 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffda365710ab499bacd697a0e42fb79e : SUCCESS in 15m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/576ae1f74292450da8f413433eae5731 : SUCCESS in 45m 21s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/bdedcde662a14fc3a9aef0fd4c899cfa : SUCCESS in 46m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfbafbb9a85741e38d976060589f422d : SUCCESS in 44m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f1290b28f8c4fa68f8349d37f60b35b : SUCCESS in 27m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1b866617794841d989cf484e2714aa3c : SUCCESS in 41m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c71342c18a9a4bd1a4839b59d5e1b101 : SUCCESS in 44m 08s","accounts_in_message":[],"_revision_number":1},{"id":"deff52c3b90c0b60767148d167d91278ba95bd5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 22:49:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30f4254277775c5cc32537c57bb756fbdad382bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 23:39:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0e9befa9bc4f738346d5118a0a0238\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb34c4d0e3ed4e9f87267b3062dd5716 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be4d857522c34d8cba01774ef06782e9 : SUCCESS in 10m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/764ed948701049e2aa43681561ed4646 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81e2e8ac18e4c6f8eb1ba93ceff6af9 : SUCCESS in 6m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b1ec7c71b3f946fa94c288bdf665375d : SUCCESS in 31m 29s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e80ec47574db4d5a811de9cd0cbff903 : SUCCESS in 25m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7e98398a33846d5aabf402bb4744184 : SUCCESS in 30m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e075e658953404284d69508d3146373 : SUCCESS in 23m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/056e3df8f7ba4261942705db2cd3226b : SUCCESS in 44m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f3ece4b9f926402db8a742f12288c772 : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8fc2352d8be7f54487eae29060dc65ea67112714","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 23:39:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"376a5f3d86832e6b63cd49c5f225a9dcbbf0c2eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 23:40:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63975d5a0a7491cbaadca9109ba2f2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b3c4805adb4fa4ae7908fb7ae41685 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6e5ec8a650a855c4e5debab5925622ee4ad317f","revisions":{"f6e5ec8a650a855c4e5debab5925622ee4ad317f":{"kind":"REWORK","_number":1,"created":"2025-02-11 11:40:58.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/941232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/941232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/941232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/941232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/941232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/941232/1"}}},"commit":{"parents":[{"commit":"b44114e1eb603c25a2b5116ff4378d1fdc953be7","subject":"Stop using eventlet in scheduler tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b44114e1eb603c25a2b5116ff4378d1fdc953be7"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-11 11:40:06.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-11 11:40:06.000000000","tz":60},"subject":"Use threading instead of eventlet for launcher tests","message":"Use threading instead of eventlet for launcher tests\n\nGet rid of eventlet when doing launcher tests\n\nChange-Id: I664c07aadb7b6fcd00c74866dd50b4662fe18206\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f6e5ec8a650a855c4e5debab5925622ee4ad317f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f6e5ec8a650a855c4e5debab5925622ee4ad317f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
