)]}'
{"id":"openstack%2Fmistral~941608","triplet_id":"openstack%2Fmistral~master~Ifcc485a2c164b8c7cf894135b5f9e83936ca60e2","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-14 12:40:03.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ifcc485a2c164b8c7cf894135b5f9e83936ca60e2","subject":"Fix the MISTRAL ascii art","status":"MERGED","created":"2025-02-13 22:34:59.000000000","updated":"2025-02-15 11:51:08.000000000","submitted":"2025-02-15 11:49:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941608","meta_rev_id":"0775faf9a870a6636caa686866d3bc7f13259ffe","_number":941608,"virtual_id_number":941608,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-15 11:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-14 09:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-14 17:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-14 17:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 23:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 09:21:44.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"48ab652ce7cd9c256cb0341d4d07c9c6a32b59bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-13 22:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3975cc8faf109274909a460a64574ab11360e5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 23:37:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d16745ba351436b9daf444a985aeba3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b813714b504da3938ae91bb6f5d450 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5982ae463fba419b8516029c78bbd568 : SUCCESS in 11m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f72d5384554b434db0773fa31e79d5fa : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04676359fce484396431114830fd643 : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cf7de5a9ede41859a0cb2e00827d7e9 : FAILURE in 19m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cb5c3135265343bc9f82babfbb202dab : SUCCESS in 32m 41s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/61ba423ab7904a6a95f281e34d8ddcc4 : SUCCESS in 37m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83b297532f174587ad89fe974f54b50e : SUCCESS in 44m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/654ef4a120884f9fb0a425c28b356583 : SUCCESS in 39m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/96512e020c08459f86efc51855a00444 : SUCCESS in 31m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7d8e4a884269422b85d85453bf8f8538 : FAILURE in 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7415d71e9488c7ff41ed9d5805661e13d88e6af2","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-14 09:21:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e99c3f9b964a4987222a1544bc38187188946114","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-14 12:40:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0af32c2a6df36994a933c23a8d046a527f6e29d0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-14 17:51:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee6e7670a090956e10150b67a0740dc792d45043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 19:02:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/444d9a820d3c49edb8d347011ab79c0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8e7083c4d547c8abd51489b7563167 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fdb04426938484fa61d5391713cf6a3 : SUCCESS in 13m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2fd10c29bcbe4927982f4038affa09c3 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5c29613e2f4532b0f4c1370da5fa9d : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d505bcacead24114848418bd66c8445b : FAILURE in 17m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/763d3b02a35e42929084903e2c73fbe5 : SUCCESS in 40m 57s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/2d6aba4807b54164a2b3e46d4a07381d : SUCCESS in 36m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a8d9e1d69cd4befb5efe771cfeff692 : SUCCESS in 32m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eb1890717b14043b582a027507327a1 : SUCCESS in 1h 02m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9e94708e35f043b89fb722d7e5e16029 : SUCCESS in 44m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6ce5c22825d94b55bbe9267387f4c523 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"305c98f92db689ad9319df5b4b099f231d1ecc66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:06:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45183182eac9f5796f495e19cdbe131feab9efc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:49:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff6f82009e5b40b393c098a8e2f60f14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef1fb0db6924728b3bb40de43bf7d29 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c87f26bb0c4796b727b7ce4b5addbe : SUCCESS in 14m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00bf8cf7bb3c4ead922e300356c93f2b : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0145d0fd520a43e5af47cc39f5972802 : SUCCESS in 6m 32s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f28cb31fc7fa46b19131faa66ef09b27 : SUCCESS in 36m 46s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/c919fc6baba045b09a9aaaa1b050a388 : SUCCESS in 34m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9e4ce8425ba4fad915c9582317f50c9 : SUCCESS in 36m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3b992e87bf941e1b2ab7a82d1f49d2c : SUCCESS in 36m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c7695dd0720247489ad7e646e62ffe4a : SUCCESS in 29m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f458dc5ed38b4c03a91e9357d63088a4 : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a65012103db21af0b76b818f30d4444b617ab0b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:49:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0775faf9a870a6636caa686866d3bc7f13259ffe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:51:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2088975074e64ced83229f8c9417a92e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f6f96d0eaf41ed868c80cd2943ac7c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcbf1e7d58be483c71368d067b946ad64a2885a7","revisions":{"fcbf1e7d58be483c71368d067b946ad64a2885a7":{"kind":"REWORK","_number":1,"created":"2025-02-13 22:34:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/08/941608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/08/941608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/941608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/941608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/941608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/08/941608/1"}}},"commit":{"parents":[{"commit":"8df35dc9feffb68cb6050f488bf68c02dbd5bd97","subject":"Add missing oslo.service.service namespace to config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8df35dc9feffb68cb6050f488bf68c02dbd5bd97"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-12 15:42:06.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-13 22:34:37.000000000","tz":60},"subject":"Fix the MISTRAL ascii art","message":"Fix the MISTRAL ascii art\n\nLet the mistral ascii art be beautiful again!\nThis is the most important commit of the day, maybe of the year...\n\nChange-Id: Ifcc485a2c164b8c7cf894135b5f9e83936ca60e2\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcbf1e7d58be483c71368d067b946ad64a2885a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcbf1e7d58be483c71368d067b946ad64a2885a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
