)]}'
{"id":"openstack%2Fmistral~941609","triplet_id":"openstack%2Fmistral~master~If3fb394dfe23e668aa6d0a759790c4b871730a18","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-14 17:51:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"If3fb394dfe23e668aa6d0a759790c4b871730a18","subject":"Stop using eventlet to start services in unit tests","status":"MERGED","created":"2025-02-13 22:34:59.000000000","updated":"2025-02-15 11:57:26.000000000","submitted":"2025-02-15 11:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941609-eventlet-removal","meta_rev_id":"2f48ff063c77c3685befccc04a19e821be509594","_number":941609,"virtual_id_number":941609,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-15 11:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-02-14 13:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-14 17:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-14 17:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 23:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 13:45:33.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"7d7f588d60cf6742cfaa5744005732c3a907c5b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-13 22:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8f6334c63ea91a9ef720342e09ff8272d7f00be","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-13 22:35:27.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"3f70fda40f99f323e8f5f3eb0a5d3363855497a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 23:26:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b892468592ce44ec9329c1e26331c5f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53620e135f3e4247a0f8ca24454a2aad : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7882d4fa51654887b3e148793868f126 : FAILURE in 15m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d51e767979d54bc18a6b349b93ecfc8f : FAILURE in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2092edf60d0b41d1ace120413a928846 : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aa245cdf4584ac1b851c5bb129840f2 : FAILURE in 14m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8714bbf0e01a43eb9ad267c9f13cb160 : SUCCESS in 38m 42s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/3c555dc89fff4616b316f34c8edf5b37 : SUCCESS in 28m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0485e152df8f4aa8b5bec55cfcfe8c40 : SUCCESS in 25m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3db229ec3d74fc497f394fa786e9553 : SUCCESS in 38m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/505e1a914016452c88b1839250c3f619 : SUCCESS in 44m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c4eba943a7df47d98f798cf17170bf90 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5180b95a35c138e8cc33f330ac649530a312691b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-14 12:39:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"471f2628baf2c4e9a1db2c2e9d29435668898c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 13:22:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e47f320917144ccf9fcf851a71a68bda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4544968e69a14a08afb791847465ae39 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/287bddfcfc384d39b00e61f3fb0e2619 : SUCCESS in 11m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af7dc3fddeda473fa00fee324c2c0b97 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a15669a0f894d5f94e80a430950933a : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/330c6a316dae45e9bb634e0aa33d8c5e : FAILURE in 17m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/452e91aadc804d488f00d3d820409b8b : SUCCESS in 24m 36s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/fc7a2640c1d94946bdb80f93efa05ae1 : SUCCESS in 24m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dbeaf9731ca41b2afa1e458caf788f5 : SUCCESS in 42m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a410f332549b4a389a50e2958e133b6d : SUCCESS in 22m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/82ceefa220d4402abdaf87f2a939a6e4 : SUCCESS in 26m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b4dfa648bec649d5a73f9687bf7d4d5c : SUCCESS in 25m 25s","accounts_in_message":[],"_revision_number":2},{"id":"77a20425d6123369b52a3278424b3ad7e88ea336","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-02-14 13:45:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f723ae6c1043c4d3e1cf1ebd2a2e2abe116729bd","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-14 17:51:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6ad92295a202bd2b25314b1beb043c261fbe763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:06:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80d35bffc4829879a7db5752a7c50709fe827cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:56:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d67eccf6964292a1bf458cb9956462\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1859477d32b24c8980b5615c195b3cab : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1720a800a82841f1b4605739a562374a : SUCCESS in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02bab87099cc493089d526eeb1ba59a9 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4183bf1ec80a48f8be344129dd295d42 : SUCCESS in 4m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3d8e5d392a7b4346a5e63220401389c6 : SUCCESS in 30m 34s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/646af82c3849424da4190d60b727a748 : SUCCESS in 30m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/566b6b42b29949b8b79f954fce958913 : SUCCESS in 27m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6890a357b1c34b5a87d4e2067d0c24b9 : SUCCESS in 29m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c1567f3215734c8eb0b7c3025119d367 : SUCCESS in 27m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9805bad17ac44797a6c3557996fb1238 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":2},{"id":"796c79c81b2a3cd4f3c860e88e898310c4afc8d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:56:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2f48ff063c77c3685befccc04a19e821be509594","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:57:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4741562aaae34cc0be4a5e7b4eaf039f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e717de6ab1bc4440b62ee1f5cd29801d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c14c59248b3f873f0b75a868d158f57db792bc5d","revisions":{"29100363176c9534893cd6f42ae2fd0aad3682f6":{"kind":"REWORK","_number":1,"created":"2025-02-13 22:34:59.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/09/941609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/09/941609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/941609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/941609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/941609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/09/941609/1"}}},"commit":{"parents":[{"commit":"fcbf1e7d58be483c71368d067b946ad64a2885a7","subject":"Fix the MISTRAL ascii art","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcbf1e7d58be483c71368d067b946ad64a2885a7"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-13 22:33:51.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-13 22:34:37.000000000","tz":60},"subject":"Stop using eventlet to start services in unit tests","message":"Stop using eventlet to start services in unit tests\n\nServices can be started directly using oslo_service, no need to\nencapsulate them into eventlet threads or any other thread.\n\nChange-Id: If3fb394dfe23e668aa6d0a759790c4b871730a18\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/29100363176c9534893cd6f42ae2fd0aad3682f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/29100363176c9534893cd6f42ae2fd0aad3682f6"}]},"branch":"refs/heads/master"},"c14c59248b3f873f0b75a868d158f57db792bc5d":{"kind":"REWORK","_number":2,"created":"2025-02-14 12:39:37.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/09/941609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/09/941609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/941609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/941609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/941609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/09/941609/2"}}},"commit":{"parents":[{"commit":"fcbf1e7d58be483c71368d067b946ad64a2885a7","subject":"Fix the MISTRAL ascii art","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcbf1e7d58be483c71368d067b946ad64a2885a7"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-13 22:33:51.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-14 12:39:24.000000000","tz":60},"subject":"Stop using eventlet to start services in unit tests","message":"Stop using eventlet to start services in unit tests\n\nServices can be started directly using oslo_service, no need to\nencapsulate them into eventlet threads or any other thread.\n\nChange-Id: If3fb394dfe23e668aa6d0a759790c4b871730a18\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c14c59248b3f873f0b75a868d158f57db792bc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c14c59248b3f873f0b75a868d158f57db792bc5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
