)]}'
{"id":"openstack%2Fmistral~941784","triplet_id":"openstack%2Fmistral~master~Ia5b21793a3ae5ec54a04886ce8b116ff9097a05d","project":"openstack/mistral","branch":"master","topic":"pep8-skip-install","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-15 15:53:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ia5b21793a3ae5ec54a04886ce8b116ff9097a05d","subject":"Skip installation to speed up pep8","status":"MERGED","created":"2025-02-15 10:24:44.000000000","updated":"2025-02-15 16:45:23.000000000","submitted":"2025-02-15 16:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941784-pep8-skip-install","meta_rev_id":"7cd271948aa58b29950a5dcf03f18be9e0022b85","_number":941784,"virtual_id_number":941784,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-15 16:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-15 12:00:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-15 15:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-15 15:53:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-15 11:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-15 12:00:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"95679b2e79ac11ce20bb65a33790f3b4d11fdd6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 10:24:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae70d73e9ce3b2577d19cf5dd4d86603acec9891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:11:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/271c44cdc58947c4a5e8144e0bc6d6d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9061fd95ea2c4dcb81bd830c6e44556e : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d19bd31d4a4439ba0134a0f4d877087 : SUCCESS in 13m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e4f3cf9ec42495f8455b9cef2c9c21c : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aaca7b6656d4843ac9137ca46c63bc9 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f30a7338e27400395f7aaf75d688dee : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bd7b68d31ff4c2eab55681b59832bc6 : FAILURE in 18m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6ab8a7e7cec3499a863b59c2ebc08d7a : SUCCESS in 39m 55s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e79ea7dc80af4aa28b2610e87d281e6d : SUCCESS in 41m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea37234a7b8a49d6b28f3d988f0387f6 : SUCCESS in 38m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57b4204b791348c98055fca391233f03 : SUCCESS in 38m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/912f588e224c40698387a461331e317b : SUCCESS in 36m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ce2ff41c608a4be0a3ef6be19188d773 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"10f8feba537cb87c60e630837b14a95e3d1fb6e9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 12:00:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"106edcd5519b1cca9b86253a5b59f82b8a2bd246","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 15:53:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d302e7be8e08ad6a403f035fd27bf079ae04fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 15:53:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37515c86348bf722c52a370a1e98128febb0f80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:44:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06182ef8070f4d14bcd0b0c22111bc64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f631145e0f84b0a9732ec08c0cb36c9 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/970b70194cf9478ba99d47030a442afb : SUCCESS in 9m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91f6bf86bbd549848beb90c129249feb : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2314dc77b1af40129891fcfc28e93ac2 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7daa3800afd4274a2df14244d912de7 : SUCCESS in 3m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d03008a471034a1e9b87cb28a9609d73 : SUCCESS in 30m 13s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/bd94734bf27e4d7fa61cf5e96f993bda : SUCCESS in 43m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a72bb85c045d4cc1abef1ab2852e4fae : SUCCESS in 47m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74753c63028c40a6829062c31e44a49d : SUCCESS in 22m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f162a1d54ca04cbd9a5887d8687c5804 : SUCCESS in 45m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b6f3be6fc8fd44d4b4357b9ba95ed23a : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e66ab2061b382e174770273c951e314c6c12a39b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:44:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7cd271948aa58b29950a5dcf03f18be9e0022b85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:45:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a9b9dffaab4f5abcdcaa8c8d9e9320\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4edbab830ced44a1ae21511c7eefa46b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a5d4dd79fd94d62be126588cbefcaee : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55bf52591501437a27702c9747d8bc8c8cb4b566","revisions":{"55bf52591501437a27702c9747d8bc8c8cb4b566":{"kind":"REWORK","_number":1,"created":"2025-02-15 10:24:44.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/84/941784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/84/941784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/84/941784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/84/941784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/84/941784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/84/941784/1"}}},"commit":{"parents":[{"commit":"6df18aa23002a4363a32a2fff6120b6565ddf5ee","subject":"Use threading instead of eventlet in race condition tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6df18aa23002a4363a32a2fff6120b6565ddf5ee"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-15 10:22:20.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-15 10:24:33.000000000","tz":60},"subject":"Skip installation to speed up pep8","message":"Skip installation to speed up pep8\n\nChange-Id: Ia5b21793a3ae5ec54a04886ce8b116ff9097a05d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55bf52591501437a27702c9747d8bc8c8cb4b566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55bf52591501437a27702c9747d8bc8c8cb4b566"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
