)]}'
{"id":"openstack%2Fmistral~941785","triplet_id":"openstack%2Fmistral~master~I7cef1e8e759071cde4592b64d59a6fd39e412937","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-02-15 20:21:49.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-15 20:22:49.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I7cef1e8e759071cde4592b64d59a6fd39e412937","subject":"Get rid of _th_loc_storage in mistral utils","status":"MERGED","created":"2025-02-15 10:26:56.000000000","updated":"2025-02-15 22:08:40.000000000","submitted":"2025-02-15 22:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941785","meta_rev_id":"7711127c7a9b528785d72cedc717c6defe8bbde8","_number":941785,"virtual_id_number":941785,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-15 22:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-15 18:14:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-15 19:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-15 19:35:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-15 11:17:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-15 18:14:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"c7d21de184e797b164676e4620ba8a25cdfeb1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 10:26:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e637777ee124eb22a1cda86b133d5b843abfa541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 11:17:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43aed5d2824d4b1bbdf3245651a4ea2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58421275ada6401f9a04ad2ec1de3cdd : FAILURE in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ed2dab6018945d4a66705a368507c90 : SUCCESS in 13m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3f3e0c6fa504734b36fdc08ab5181cb : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440590340ed7426591a18a42313741b2 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4676ff2e44834e1e921d11654b25e8b2 : FAILURE in 18m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/12c977771b0b4fed913faa69cbfc321d : SUCCESS in 31m 51s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/aac267ae52b04b89bd7c4367116a55e4 : SUCCESS in 29m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9cf69bdb41045ada22c8b947aedb279 : SUCCESS in 37m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f512cd794f4e442ca568fae207e632ec : SUCCESS in 29m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e7128f3339304b50b57158fa0624ac4f : SUCCESS in 24m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b82c13f2208b46d8b22d2c7d5f0e38ce : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e8e7cff551a68003caf0752684a16f54be59b45b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 18:14:13.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3d92f4d21692146813cc1fd909a462d69fbc2623","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 18:14:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e621d9a09434cb35ff1ce28238d16435077a1d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 18:54:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf696ad6cd48483782e58d805f12a6ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa324a72cf844f078c14fa038d793a52 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecdc2dea016c47edae6183b50b9d25ac : FAILURE in 14m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a92a82e8b3c47e6a56f30c504092815 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d851f02e54c64ded9ec38bd8f457988b : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cca74951adb4458b96b72b726854b4ac : FAILURE in 16m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1c3bf5df376d4638a745f70f791f966e : SUCCESS in 40m 02s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8bc5e831c557411fa2cb97399516c6ed : SUCCESS in 39m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a84defeef914169bc721747fc5383c1 : SUCCESS in 23m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5a98fdb02dc4a2cadeded895299cbe6 : FAILURE in 10m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/04257fb95a5d43448c96b91184c99345 : SUCCESS in 21m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/82bfb35a42c94094a856da5937ac1bf5 : SUCCESS in 24m 22s","accounts_in_message":[],"_revision_number":2},{"id":"725d9ba0eaf31bc601ea6539f2523a68a7da8d31","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 19:02:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"faa6772036a16e8ac060e880c24921e546d47d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 19:30:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f2583deca043d184f8eda27cbc8b42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67b7a57dd604ef4918e5bd035bda2e9 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ddea989c85f4d31ad8bb9ee52830c24 : SUCCESS in 9m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a0b9ee5e98441b3ac4816163bcdb9cd : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa81f16e083845e5b5e47bba02cf98be : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/983b22d7c35644c980a54b4810bc24b0 : SUCCESS in 14m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/64e849e2d5ce4d7c9635844c9df66b4a : SUCCESS in 24m 34s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/b6406ef74f38413f90853fdf667ccb81 : SUCCESS in 22m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dfeae76cc7846eda76abc7553f872b1 : SUCCESS in 26m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc665ca63de747629d2c8b66b85d10f4 : SUCCESS in 24m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/11849d3e02d942609153c3ec0ac29eec : SUCCESS in 24m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/321b6aec1d2c425980e40c6238b7b5a9 : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":2},{"id":"014a3e5dfa955b0c4c35aac1eea21f427016782c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 19:35:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67ea27f44be2db3954d3c244826a1f4352268991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 19:35:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96f8b6acecac0949f32eeae73da74a98b70ff178","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 20:19:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4993a45f55dc409ca784d68fa6229028\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613a0a44c6a94b5598a020f03e25dbed : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14eb9db6fb3944629c731814618e91b6 : SUCCESS in 12m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bacd2b8dbe70473ea3f58fe556f3ef22 : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4dd4c054804bb6866cade69f4136bc : SUCCESS in 4m 36s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a7fa1c5122994d71a5fc490258c06675 : SUCCESS in 44m 09s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e271a2c5a9814ba1b14417394d6dd254 : SUCCESS in 40m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7931e69755294951bc850378b92fa907 : SUCCESS in 23m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30cd363b9ac448d4a8e8244fee1dc785 : SUCCESS in 42m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/932fb75f3676482eadce7ccc7117f2d4 : SUCCESS in 43m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9a73e556ee634bd68b2c4d9b67a83dc6 : FAILURE in 12m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a474c10ffda4c85a0de824da7dc9d5615c680335","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 20:21:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9e16a31beb325e31f0200a31a016b39f6622ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 21:11:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503e716325d0493e8722a2ae9ceaa151\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17fb06cb99c7419696b34709668f9c7f : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/599599c00be34c1fba889f3457920348 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a6c17e1e3504f66bad58582adcbcf9c : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cde4504bb984582bef5182c63e08eaf : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caf2b1175ba54229ae7718659d3fba7d : FAILURE in 16m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e054244bfbdf4db991cf165fb10059e7 : SUCCESS in 42m 59s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5c162d91107a40bb9c787d571b645729 : SUCCESS in 24m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2bbdf0309d0482bbe8549e2c3e3295d : SUCCESS in 44m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4100626873b4e57ad63d9edbb01396c : SUCCESS in 28m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0ee80217f89a4cab8e8754a842248164 : SUCCESS in 26m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/51d0afc75b4e476a9e1f19adaf1fad84 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4c96dab25075d3c3161ec876836705d3dfea8f0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 21:12:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd7d319ca3f5d58392a8a209ba57a3dfecfcd0d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 22:07:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e34dddf12fc461eb7934713994f1a8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684dcf8e69454012b0a324ba6b585f28 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e093f9dc634d4993b9bc2d6e93e89a81 : SUCCESS in 12m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/353f59d1cdb4489fb920bd466a66af99 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af887ca14a05483f841fbeabea8233cc : SUCCESS in 6m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cf456cea429544b48b36a5b8a5a300f1 : SUCCESS in 45m 31s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/3aa2f19a63614ed9ad82511efc2305e3 : SUCCESS in 55m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af34942e6eb642679e5c48dc0aa0e0d9 : SUCCESS in 25m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e0cd3c7e50040a39e17ab0e6d8e059a : SUCCESS in 24m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7eb1b08f00184e58b42426d1018b4411 : SUCCESS in 23m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e6f5f48d1d0f4eb3a32ac20d10d5dfd9 : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":2},{"id":"04883c3f2534d4518c8e9a4d32735732b432303b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 22:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7711127c7a9b528785d72cedc717c6defe8bbde8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 22:08:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f211e087e697443182ae352f67b49cbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc83181531e34af8b834672751c2d554 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff2c2b3c38c5133cdeaa568a0b8f5d2ad5cf8f63","revisions":{"91085273907f58cc80600af4c3d32c0c719cd16e":{"kind":"REWORK","_number":1,"created":"2025-02-15 10:26:56.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/941785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/941785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/941785/1"}}},"commit":{"parents":[{"commit":"6df18aa23002a4363a32a2fff6120b6565ddf5ee","subject":"Use threading instead of eventlet in race condition tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6df18aa23002a4363a32a2fff6120b6565ddf5ee"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-14 18:13:28.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-15 10:26:44.000000000","tz":60},"subject":"Get rid of _th_loc_storage in mistral utils","message":"Get rid of _th_loc_storage in mistral utils\n\nThis is not used anymore, it seems to be a leftover from a previous\ncommit\n\nChange-Id: I7cef1e8e759071cde4592b64d59a6fd39e412937\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/91085273907f58cc80600af4c3d32c0c719cd16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/91085273907f58cc80600af4c3d32c0c719cd16e"}]},"branch":"refs/heads/master"},"ff2c2b3c38c5133cdeaa568a0b8f5d2ad5cf8f63":{"kind":"REWORK","_number":2,"created":"2025-02-15 18:14:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/941785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/941785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/941785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/941785/2"}}},"commit":{"parents":[{"commit":"6df18aa23002a4363a32a2fff6120b6565ddf5ee","subject":"Use threading instead of eventlet in race condition tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6df18aa23002a4363a32a2fff6120b6565ddf5ee"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-14 18:13:28.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-15 18:14:13.000000000","tz":0},"subject":"Get rid of _th_loc_storage in mistral utils","message":"Get rid of _th_loc_storage in mistral utils\n\nThis is not used anymore, it seems to be a leftover from a previous\ncommit\n\nChange-Id: I7cef1e8e759071cde4592b64d59a6fd39e412937\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff2c2b3c38c5133cdeaa568a0b8f5d2ad5cf8f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff2c2b3c38c5133cdeaa568a0b8f5d2ad5cf8f63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
