)]}'
{"id":"openstack%2Fmistral~942398","triplet_id":"openstack%2Fmistral~master~Ie1b5ab52ff0f65e4c772bed04c30d2bb35dc623f","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-21 10:59:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ie1b5ab52ff0f65e4c772bed04c30d2bb35dc623f","subject":"Add some logs when starting unit test executor","status":"MERGED","created":"2025-02-20 20:45:01.000000000","updated":"2025-02-21 11:57:18.000000000","submitted":"2025-02-21 11:56:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942398","meta_rev_id":"5b587be9942c9a2cbd1727e811ff038014708a4f","_number":942398,"virtual_id_number":942398,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 11:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-20 20:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-21 10:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-21 10:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 20:56:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-20 21:20:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"114ec87cf4bf8809b1645f249bc4633c664b2836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-20 20:45:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"898c93d35d63f7b6b2269cc285890f11c08e5b5c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-20 20:56:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ec68d92dcd1cf39e6766db304754341d0098e220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:20:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/179f1e2f0bcf4f90b7d5d2137e89814a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3ab0454170464ba661618ab65f5d3c : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41da28263c17497ab21c2f7328e25368 : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e585721cfefb439fb207d6ecd35f526d : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15321cd82ea1449eaaf7658c23f173b1 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fabc3ecd6aa42ada3ab82dfbe3b0e80 : FAILURE in 16m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a94e6b1a0a904d4085e26e65b3709e95 : SUCCESS in 26m 26s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/a57dad59b4464f5d936f0d96288be265 : SUCCESS in 26m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b682128bcd34e5a93a1ed2cc00e899e : SUCCESS in 26m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f88085ae60724929bc168e0ffec1ff0f : SUCCESS in 32m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bf68f19ca5dc4584940eadd55ba59ce7 : SUCCESS in 25m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2f9a939957244fdaa8b3071c1a550442 : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"00acc1e187698e66440dd11444e4c2f7731ddb1f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-21 10:59:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e107886f60542d3299f9818d2daff6f6818da2c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:00:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6cda3f006722972039e0d5ec9c1612a565ef6ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:56:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b89d84b2c2f4635a1f585e339346c8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcad9225013049dab544905803e1eb11 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c74d70bd75143cabd74c9b2d8fbeb8d : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b690ccb297d446999f7e685ed994c7db : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a6d547397e456596fdc7729f1525dd : SUCCESS in 7m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fd6056660b8749dab8d69ecdc2588a94 : SUCCESS in 28m 20s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5a6769f170544d0397caed875bc082ff : SUCCESS in 47m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e40a2bdad364826b9701c55acf96935 : SUCCESS in 25m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da419c6e36341fd8d0541229759a3f8 : SUCCESS in 42m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ed91f7026bf04fdb9557385e3b4ad14a : SUCCESS in 50m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d7e3d78cfa4f439c927509c90a695c0a : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0d2c7c1e2f94dfa90270f21bd69eb66c80c049f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:56:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b587be9942c9a2cbd1727e811ff038014708a4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:57:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4fe63abe1f840c2ad656058c267e14e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ac885cfcb94379aac100e715a12bef : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"168f375597a8b1d432f228a8342e40c5efec3902","revisions":{"168f375597a8b1d432f228a8342e40c5efec3902":{"kind":"REWORK","_number":1,"created":"2025-02-20 20:45:01.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/98/942398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/942398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/942398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/942398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/942398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/942398/1"}}},"commit":{"parents":[{"commit":"6311d3de404bb513edc05c9f48071382553a7858","subject":"Simplify the TimeoutThreadWithException class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6311d3de404bb513edc05c9f48071382553a7858"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-20 14:52:45.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-20 20:44:46.000000000","tz":60},"subject":"Add some logs when starting unit test executor","message":"Add some logs when starting unit test executor\n\nChange-Id: Ie1b5ab52ff0f65e4c772bed04c30d2bb35dc623f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/168f375597a8b1d432f228a8342e40c5efec3902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/168f375597a8b1d432f228a8342e40c5efec3902"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
