)]}'
{"id":"openstack%2Fmistral~942405","triplet_id":"openstack%2Fmistral~master~I56b212013e3beff78d458c51c0297a7a4f0f742e","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-21 10:59:34.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I56b212013e3beff78d458c51c0297a7a4f0f742e","subject":"Drop eventlet to spawn sqlite test","status":"MERGED","created":"2025-02-20 23:38:05.000000000","updated":"2025-02-21 11:44:24.000000000","submitted":"2025-02-21 11:43:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942405-eventlet-removal","meta_rev_id":"fae1b89bd3cbfe05cfa3888b4e4a1e2095fde6cf","_number":942405,"virtual_id_number":942405,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 11:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-21 07:45:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-21 10:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-21 10:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 00:26:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 07:45:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"fb4401ceeb338934b79d8ae9b978330c7b7d6fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-20 23:38:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07a8913c7aae32ab0f875a4b36e685b847f1103d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 00:26:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e8df97343a40429f03063eae42813f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c402eee277146b9b7394f535a1f3827 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7935f8f59b2d40fe8a4f7e3f0e65cc74 : SUCCESS in 13m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83c5d6dd11764c168891459ac32ff194 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d4f08e40ee45bcb38a99f56a6111f8 : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b78ad4f3924f4a7cb88c25d79fbf821f : FAILURE in 15m 20s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a94db47aece5412f841ec873880d2dbe : SUCCESS in 43m 02s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/2f9a15a787be4f6089cc6905f821a9c5 : SUCCESS in 21m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68853f4534cf460dae67d3f78b642277 : SUCCESS in 26m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49845f7760804509b7f028faaee7b620 : SUCCESS in 26m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ccf20d9a73c74b788a3f5670c37ccb29 : SUCCESS in 24m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e2f04e8bf17e48d8b258bce6d9c6bf90 : SUCCESS in 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1c8262a1f3be29353f85f56190e062d38418ead9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-21 07:45:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80be805af529a189f359ba5e7dc8134176d97328","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-21 10:59:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1be2224e611a3b86e701a3f1fd6ddefa9209fcc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:00:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65bf264c26c4fc71b3e08f4c45d78f446255a621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:43:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbecd752d9b84101874a6be996ab9601\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4644ce8d61441ea2162c06148e7879 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd220e34ed14b208ea46fe05e17f8f1 : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fc76390800546899e4bb4c70c5c8f35 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5248be61ef274e09947225ac88b1014c : SUCCESS in 6m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dde4b24e30e34dce83911b040c7c3a70 : SUCCESS in 27m 03s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/70e618ab5dae48cb8dbc599d0883af35 : SUCCESS in 28m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0612c79c81804d69b21f4153e5aa619f : SUCCESS in 39m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fec951c0c857436293820aaaf430776a : SUCCESS in 35m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/77fd41b203e445f1a701df8f00605c28 : SUCCESS in 35m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/66c418f54c3c47ce98a98ba0d9948ce7 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":1},{"id":"02e5ed46bea638eb8cbfb50c1b08602b3a52841d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:43:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fae1b89bd3cbfe05cfa3888b4e4a1e2095fde6cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:44:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f37d6abe5d4830ad9f7676807abc77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2900ee8f503a43518000426b634302bc : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2de1de10b7cee1a13123f3430f8c3eaa5525366","revisions":{"d2de1de10b7cee1a13123f3430f8c3eaa5525366":{"kind":"REWORK","_number":1,"created":"2025-02-20 23:38:05.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/942405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/942405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/942405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/942405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/942405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/942405/1"}}},"commit":{"parents":[{"commit":"6311d3de404bb513edc05c9f48071382553a7858","subject":"Simplify the TimeoutThreadWithException class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6311d3de404bb513edc05c9f48071382553a7858"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-20 23:36:59.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-20 23:37:56.000000000","tz":60},"subject":"Drop eventlet to spawn sqlite test","message":"Drop eventlet to spawn sqlite test\n\nUse threading instead of eventlet\n\nChange-Id: I56b212013e3beff78d458c51c0297a7a4f0f742e\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d2de1de10b7cee1a13123f3430f8c3eaa5525366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d2de1de10b7cee1a13123f3430f8c3eaa5525366"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
