)]}'
{"id":"openstack%2Fmistral~942406","triplet_id":"openstack%2Fmistral~master~I39b155ab9db0fd7e963fe4faebf93b76810cb08a","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-21 10:59:45.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I39b155ab9db0fd7e963fe4faebf93b76810cb08a","subject":"Drop eventlet to spawn parallel tests","status":"MERGED","created":"2025-02-20 23:45:09.000000000","updated":"2025-02-21 11:54:53.000000000","submitted":"2025-02-21 11:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942406-eventlet-removal","meta_rev_id":"2574da6dc7686f95ef6b6c4582386c06d2210612","_number":942406,"virtual_id_number":942406,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 11:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-21 07:45:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-21 10:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-21 10:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 00:20:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 07:45:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"ed2ddb69e2bcf4fc15ec0f34a8d04c783a9ff7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-20 23:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f73787b799e323773d0a596825db03f77b89f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 00:20:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd1f1505f41421293e9f43cc8f6a0c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9a14942e954327864416da51b5d43d : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a00470a9d0444b44906fdff9f1b00048 : SUCCESS in 11m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c45e39b75e4247c58c8d3bec03baf7b8 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4906fbac02d7411b816250ce0670335b : SUCCESS in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76b055c50edb41648805ec7951ac0125 : FAILURE in 16m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4dea1cc9f86c44de9c407cb3949684eb : SUCCESS in 27m 51s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/4ff96afb6f494a16980f19b584eace2f : SUCCESS in 26m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d948f1929dc24c3f9ff78dcac4b0857a : SUCCESS in 27m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4340fe2f21da43fbb4f03240d669e798 : SUCCESS in 32m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/115b3d4aff124927a797782b0fa0667a : SUCCESS in 28m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cfd9748c2a1449ac8f4d74a55f43ba5c : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e8df6a9bdf06c79474a89aed7383be3d10d74aba","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-21 07:45:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2eca77dc4934d7cbeb8a115333df628fe7d24f48","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-21 10:59:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69cca1fa6a7e645ce81827b518a37546f78d3215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:00:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac29e86c8c947550a0ca3300a948e7eebf0623f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:53:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e8ca47fabb0427cab4d00e6fc44e4c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a13991ab451c496c88cf1fc7a8caf198 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e631f4669f5d4f32bbb54fabcf3576ec : SUCCESS in 13m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c3b276426f04c2eb48723aeceb43272 : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead554086b5142caa5477eb4402b05ff : SUCCESS in 3m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c1449d78058448d0adf7f493de5afe66 : SUCCESS in 24m 54s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/14dffe9044ab43dd85c1da5323ac6d2d : SUCCESS in 32m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/504e2689845e4ac1a4ae7da6682e9ab5 : SUCCESS in 36m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ce41c077f11486e90e5f58456818764 : SUCCESS in 37m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1f5b2a57cac1417d917658202fd5ceb9 : SUCCESS in 42m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4d1119ae7835404f86fd9ae674d0606e : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"523dfdab2e07e840d4101ac446b79234296b9484","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:53:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2574da6dc7686f95ef6b6c4582386c06d2210612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 11:54:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a833869229154486ab26087e91aef78e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c39d37d8a48400a8f68113aac6abfa4 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe0f449dfd30966aeb9a94f371bc7122c125cddb","revisions":{"fe0f449dfd30966aeb9a94f371bc7122c125cddb":{"kind":"REWORK","_number":1,"created":"2025-02-20 23:45:09.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/06/942406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/942406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/942406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/942406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/942406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/942406/1"}}},"commit":{"parents":[{"commit":"6311d3de404bb513edc05c9f48071382553a7858","subject":"Simplify the TimeoutThreadWithException class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6311d3de404bb513edc05c9f48071382553a7858"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-20 23:44:10.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-02-20 23:45:02.000000000","tz":60},"subject":"Drop eventlet to spawn parallel tests","message":"Drop eventlet to spawn parallel tests\n\nUse threading instead of eventlet\n\nChange-Id: I39b155ab9db0fd7e963fe4faebf93b76810cb08a\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fe0f449dfd30966aeb9a94f371bc7122c125cddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fe0f449dfd30966aeb9a94f371bc7122c125cddb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
