)]}'
{"id":"openstack%2Fmistral~943824","triplet_id":"openstack%2Fmistral~master~Ia5f8b4ac886b0133468d1aa87f1a61c2807eac20","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Ia5f8b4ac886b0133468d1aa87f1a61c2807eac20","subject":"Fix a log in action heartbeat checker","status":"MERGED","created":"2025-03-07 20:09:40.000000000","updated":"2025-03-08 07:53:10.000000000","submitted":"2025-03-08 07:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943824","meta_rev_id":"b180d0028a815edb168f1e717e7db3e49bbb0c9c","_number":943824,"virtual_id_number":943824,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-08 07:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-08 07:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-08 07:18:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 21:00:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73c03135992e83642f6c86e3efecdb9662be0cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-07 20:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"328f9ab7a46588d4a9ab3a7052dd16ed0dac9685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 21:00:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2befbfaab264f41bb89775d64dd5fce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3318c696361456498df3f5bf13ed61a : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6912d78220543f893cee3e9bf44d8af : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/757af4d9970a467cbb1d7fa45bd907b6 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ac6f41b0ef49f3bee36992b47538d7 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c296e967e8c44edb6332c4c3229bf83 : SUCCESS in 12m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a11c1472e0424ca5b7edaecd6edf99d4 : SUCCESS in 26m 17s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/ce12ee9f696b42179416fb82be7a267f : SUCCESS in 23m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0836d0a73e349bd96357668dfb86b04 : SUCCESS in 28m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5c600a7d9ef45e289f4f736584d8fb8 : SUCCESS in 17m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/378e6d25426446f1af054d76d61e0f78 : SUCCESS in 44m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/23e156b3159347918b5601545b82ca0e : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"566d2a58dd7b88ef348efa85a664b3a962f730d1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-08 07:18:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38ff55bca1c143c5cdbea3836ef12f9b8e1aa76c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 07:18:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e8b2bde4b59c71c550bdd244ebe7e96445e6160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 07:52:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e2a42c6b6b4e488d15dff6692903ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e370d0c738d046b6b7792f49260b8ba1 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5713d7e2e24a16bcc4e42c2847c12c : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d50da200ce4c4b7f9e676c7de97c3296 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482d01f605a54bc89c3aebc8937ba8e5 : SUCCESS in 7m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/046c31b09ed447c5887d3ac22ca99d22 : SUCCESS in 33m 02s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/57161579b649433c93efd5cde05810ae : SUCCESS in 23m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22d1037fd8884f28b2104b3c111e270a : SUCCESS in 31m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/530c5c18e44e4112a63dd50f80ca6690 : SUCCESS in 27m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/10eed8a9a2a942d59d649099b523ff48 : SUCCESS in 28m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8ed3779937044dcbad24bba3d092b9bc : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2c2314a3792ba5fd7935d699c7e522f794c4d961","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 07:52:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b180d0028a815edb168f1e717e7db3e49bbb0c9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 07:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a797dc6fd65b4e4c82bbab7abcec91e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fa76da32d14f4fb28a251f7752063e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9079fa11e6c5e18636e7238c2460cbf38ee5138","revisions":{"e9079fa11e6c5e18636e7238c2460cbf38ee5138":{"kind":"REWORK","_number":1,"created":"2025-03-07 20:09:40.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/943824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/943824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/943824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/943824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/943824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/943824/1"}}},"commit":{"parents":[{"commit":"c095c3715d5506e6a7bee3aa56ca1fd8cf08b551","subject":"Deprecate mistral coordination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c095c3715d5506e6a7bee3aa56ca1fd8cf08b551"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-03-07 19:57:35.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-03-07 20:09:29.000000000","tz":60},"subject":"Fix a log in action heartbeat checker","message":"Fix a log in action heartbeat checker\n\nWhen an action is timeouting, the LOG info was failing due to the fact\nthat it was not able to transform actions_exs in str.\nAnyway, let\u0027s print only the id of each action in that situation.\nAlso, print that in warning because it\u0027s not expected.\n\nChange-Id: Ia5f8b4ac886b0133468d1aa87f1a61c2807eac20\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9079fa11e6c5e18636e7238c2460cbf38ee5138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9079fa11e6c5e18636e7238c2460cbf38ee5138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
