)]}'
{"id":"openstack%2Fmistral~944015","triplet_id":"openstack%2Fmistral~master~I57cd468cba227cc8cc7960e84e9a111217e80cad","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-04-23 21:59:31.000000000","reason":"Change was submitted"},"34378":{"account":{"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},"last_update":"2025-04-23 21:59:31.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-23 21:03:27.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I57cd468cba227cc8cc7960e84e9a111217e80cad","subject":"Add allowlist and denylist for action providers and actions","status":"MERGED","created":"2025-03-11 09:56:54.000000000","updated":"2025-04-23 22:00:25.000000000","submitted":"2025-04-23 21:59:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"944015","meta_rev_id":"c13920cba16bd6c1e15d2a5f295597b2720fbb15","_number":944015,"virtual_id_number":944015,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-23 21:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-23 21:03:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-23 21:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-23 21:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-11 10:17:42.000000000","updated_by":{"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},"reviewer":{"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},"state":"REVIEWER"},{"updated":"2025-03-11 11:02:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-11 19:16:25.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-03-11 19:32:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-03-12 09:37:29.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-04-23 21:03:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"21ff6c9cc982778e5e6ed88431f3d88903f12b4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-11 09:56:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc373f966c95bb4e7e30c26371f098fdc46b1f06","author":{"_account_id":34378,"name":"Julien LE JEUNE","email":"julien.le-jeune@mailops.fr","username":"jlejeune"},"date":"2025-03-11 10:17:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f569344b2fe6d6d0e87fe7838ffa6fd22d383402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 11:02:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/209da305499c4659bcbecba1dec58f78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1122a597e74159b6a632f2727e0934 : FAILURE in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebac14e2cb9a49ae9da6de25c08467d1 : SUCCESS in 11m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58384cc6f821404d8d02d90c26b29d0a : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638ad988d9724db787e66598503a981a : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce319d64b85040a68eb0f174841839ef : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88b10c72bfef4beeb6a48a8b267adb50 : SUCCESS in 14m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f4970c27c6424d4e8d155217e9e3d756 : SUCCESS in 24m 55s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/abe902cd5c5149b7be92589230966168 : SUCCESS in 33m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48113ed0d23c49caab0a2f3d7b2b2097 : SUCCESS in 45m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6507ed9d79a841a39d61ab3d9a8ff38f : SUCCESS in 25m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/12023198f1c64df297ea35b6c2aadef2 : SUCCESS in 36m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d706387d13b047a39e487630af6a2794 : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":1},{"id":"41a0b8f3bbfdce4190f1d892bd02ef0ca0979a0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-11 12:06:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"272a50542327e70d96bd2940115dbcbbf1971f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 14:05:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aae26c36f8849198264d94835794507\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc39b3897684388ae05f9c1df43ee0c : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eef611225b0e4b0baf5ef770ce8d4487 : SUCCESS in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/150723736b394fd8976e96c4028add08 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92936d14f4554ba6a44c8a0cca6375aa : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06ce35314569437da56b1040fc5dd21c : SUCCESS in 1m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf3e821cac35451f9f792bda7c9a9af7 : SUCCESS in 12m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/03c592c3257644a6a5e497bf970d75d6 : SUCCESS in 17m 53s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/954247f5655549bf8dab92dac825b3ce : SUCCESS in 1h 11m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4c7c1a14faf4c69990c314b9ae5676f : SUCCESS in 1h 57m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a80c3ef198b4372947d79a33cccb590 : SUCCESS in 28m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a6b9d016ddb9462baab690fc7b89c858 : SUCCESS in 37m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/259325be15f3426393a2ff4f0a0a0dee : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f8106f5b875e419a9aca12a8bfed8e772cc51f11","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-11 19:16:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7fbc4fdfab90d5dfff8ef19a33cc6fe0507cd183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-12 07:37:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a6840095e80dab4ed9d878ef7c2c111513dc510a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-12 07:38:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"63d48bc3617b333a63a69130a2c42c6d7df6f3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 08:16:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a5a6f8608bf43b98437ef0c5a17d84c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a723a462b7a94f239d2f8c701b21a981 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b11e4bad04c492a9f6654c6857fd400 : SUCCESS in 9m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cb4de2135444e37941ea52531aefc8e : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bd6b0e82484340b9bcb58c1cb9969f : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21436c1dceb34b7d9221d3c9e0cab8f0 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebdc586b992f4465b057c6205ae945a4 : SUCCESS in 13m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f9009a3146642358b435f5141fdfb8c : SUCCESS in 37m 47s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/eec0487699f2492c9c811cd7dc24b18e : SUCCESS in 16m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ebfb45126c14cee887a693b85785a4c : SUCCESS in 25m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bdc93e816c0469c94b05c6b96a730dd : SUCCESS in 23m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8fd4b68b713b48048511bf14e4e8d816 : SUCCESS in 36m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/88f01c81f6044875b737e46b534d4e15 : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":3},{"id":"797494e71935a22439df03ea742c8e48969d6bfa","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-12 09:37:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6b60d155c5cb7d5a10f6837361e300664644e37","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-12 12:47:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21eb363be4e3ffb3edb56a2e38756656c8b2dbb0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-17 17:57:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"63656982d002f5fc700405fd7450fb247246ccef","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-26 22:16:12.000000000","message":"Patch Set 3: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4fd2792cb25113cf3daaa7e45f07810ee0c9d099","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-31 12:25:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c415914ee7b365bc2c9e897488f053f561579a4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-23 21:03:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"49f863cafc43116290dd64f45fdb48ce8a74b7fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:03:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"daa86db792757cde706d8f0eb8daa0d7284ce57d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-23 21:03:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3c4d1069efa09b9bf87ef7f336c6bbad7a3e0499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:59:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6ffb563e4f4352b22aff6518057ba3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349ad600a93e469a95588018f4d0db1d : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89df0d854c0f4ce48e38141f550fc6b9 : SUCCESS in 8m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a581adb02c534b958c5d367abb8106f1 : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/671df10786a44928a5d8b39928a4776a : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8138873cf00d4ad884d7cf0dff9fbbf5 : SUCCESS in 2m 46s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/79412347fc204a6f9bafd373da8ab023 : SUCCESS in 50m 30s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/80b7aaffdac74fd3becae16aa1ca4269 : SUCCESS in 15m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7beb7bb07bcd4797bb054f2316701c5b : SUCCESS in 17m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf455fafd63e4732a7c4ad130f95e636 : SUCCESS in 36m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/33f5dbf5bfce49d99c1f2d3c7678c6eb : SUCCESS in 28m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c16d2bb7c3cd41738d3cabbcad470b16 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0c62b6399dcbe200bd28fef5308f3efa17901617","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:59:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c13920cba16bd6c1e15d2a5f295597b2720fbb15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 22:00:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce1e91a0222472a9e64f09f08c72bee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8c58b4e6f84a51aba6cc78840ff908 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60a695b83d294b32b8acf11920b5759f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e","revisions":{"58e91f08de8f604781031f9c6af4591a77f442ab":{"kind":"REWORK","_number":1,"created":"2025-03-11 09:56:54.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/944015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/15/944015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/15/944015/1"}}},"commit":{"parents":[{"commit":"e9079fa11e6c5e18636e7238c2460cbf38ee5138","subject":"Fix a log in action heartbeat checker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9079fa11e6c5e18636e7238c2460cbf38ee5138"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"subject":"Add allowlist and denylist for action providers and actions","message":"Add allowlist and denylist for action providers and actions\n\nThis adds allowlist and denylist ListOpt config options for\nfiltering what action providers is loaded\n\nAlso adds allowlist and denylist ListOpt config options for\nfiltering what actions is loaded by the legacy action provider.\n\nThis is used in a security context where we need to restrict\na environment to only load actions using the legacy action\nprovider and then filter what actions is explicitly allowed\nor not allowed to be used.\n\nChange-Id: I57cd468cba227cc8cc7960e84e9a111217e80cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58e91f08de8f604781031f9c6af4591a77f442ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58e91f08de8f604781031f9c6af4591a77f442ab"}]},"branch":"refs/heads/master"},"35c38d2b8936558086098444e72d3af9a4ac7f70":{"kind":"REWORK","_number":2,"created":"2025-03-11 12:06:25.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/944015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/15/944015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/15/944015/2"}}},"commit":{"parents":[{"commit":"e9079fa11e6c5e18636e7238c2460cbf38ee5138","subject":"Fix a log in action heartbeat checker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9079fa11e6c5e18636e7238c2460cbf38ee5138"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 12:06:19.000000000","tz":60},"subject":"Add allowlist and denylist for action providers and actions","message":"Add allowlist and denylist for action providers and actions\n\nThis adds allowlist and denylist ListOpt config options for\nfiltering what action providers is loaded\n\nAlso adds allowlist and denylist ListOpt config options for\nfiltering what actions is loaded by the legacy action provider.\n\nThis is used in a security context where we need to restrict\na environment to only load actions using the legacy action\nprovider and then filter what actions is explicitly allowed\nor not allowed to be used.\n\nChange-Id: I57cd468cba227cc8cc7960e84e9a111217e80cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/35c38d2b8936558086098444e72d3af9a4ac7f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/35c38d2b8936558086098444e72d3af9a4ac7f70"}]},"branch":"refs/heads/master"},"11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e":{"kind":"REWORK","_number":3,"created":"2025-03-12 07:37:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/944015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/15/944015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/944015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/15/944015/3"}}},"commit":{"parents":[{"commit":"e9079fa11e6c5e18636e7238c2460cbf38ee5138","subject":"Fix a log in action heartbeat checker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9079fa11e6c5e18636e7238c2460cbf38ee5138"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 07:35:55.000000000","tz":60},"subject":"Add allowlist and denylist for action providers and actions","message":"Add allowlist and denylist for action providers and actions\n\nThis adds allowlist and denylist ListOpt config options for\nfiltering what action providers is loaded\n\nAlso adds allowlist and denylist ListOpt config options for\nfiltering what actions is loaded by the legacy action provider.\n\nThis is used in a security context where we need to restrict\na environment to only load actions using the legacy action\nprovider and then filter what actions is explicitly allowed\nor not allowed to be used.\n\nChange-Id: I57cd468cba227cc8cc7960e84e9a111217e80cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
