)]}'
{"id":"openstack%2Fmistral~944098","triplet_id":"openstack%2Fmistral~master~I2b874b7ab2d213a8859b7b4bbd9e546278595c0c","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-04-23 21:04:17.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-04-07 19:24:25.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-23 21:03:35.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I2b874b7ab2d213a8859b7b4bbd9e546278595c0c","subject":"Support fnmatch glob for allow/deny lists","status":"MERGED","created":"2025-03-12 08:55:18.000000000","updated":"2025-04-23 22:01:10.000000000","submitted":"2025-04-23 21:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"944098","meta_rev_id":"b055ee8ebfdf496320d2648fb6b4f1abae10d42a","_number":944098,"virtual_id_number":944098,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-23 21:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-23 21:04:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-23 21:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-23 21:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 09:37:51.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-03-12 09:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 21:29:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"cf4dcaee0b41b962273cecb0175a3839ccc29532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-12 08:55:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"321a50d89dc740562caf1af71d509a81baabb586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-12 08:59:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9a265b7223d384bf05e6ce0d048551964eafff7","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-12 09:37:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5a4d76a32b2107903a7a7b7652e8653e5112421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 09:49:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/842f68768c85482ab7bbb468f5af3c91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45628df41c54d889dd12d201a7b77ed : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ead8bf18dec34124b61ec009d8115545 : FAILURE in 8m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/358b2d13722b43c5a0e785fddeeb65f7 : FAILURE in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4b2f78512b41d7ad1cd7b2c15408a4 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cc83a5cfa214229b55afe1c12593e29 : FAILURE in 18m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5ce82bfc18c4f71981b02efe476f0a1 : SUCCESS in 28m 18s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/36c81926261f4c8ba5250d8f15c00f02 : SUCCESS in 20m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3556ee09dbb742eba081972da87993f6 : SUCCESS in 44m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efb1ae1ef30c4db18511d1c48fe8e41a : SUCCESS in 29m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d1de1371b9e54066bdaa1acc74a8da2b : SUCCESS in 37m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8dbc523fa81d4fb3b853c625d0715a55 : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3a6676356e98125f3789e1f8d6464cfba66629c8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-13 21:29:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d69c3c19f1d969c854186847b538b5285d4b0f42","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-13 21:30:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2dd09c181d4ca5a16310a83c56d51afd658d08c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 22:07:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c922abc8b944a9ab37580c7d6e447d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f629fde3e8843fca929f456b846c936 : SUCCESS in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb37ad0510ac4bcfad11d77547183b84 : FAILURE in 9m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef4b339fc0ae4c39a15223dbb2f0ba5f : FAILURE in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5729d970389495f87596fd5c60659bd : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be57284c6a714e9e8c887fed73dead41 : FAILURE in 13m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/82c8e28426524d2eb2df070a72a0cf70 : FAILURE in 17m 40s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/0802ee02bf964be0950abf3d18a062b4 : SUCCESS in 22m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aac20e318e24dc4b195aad5f9a0e20d : SUCCESS in 31m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f480480ec786444d8c40feaf1f5ef7d7 : SUCCESS in 18m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/144762a1bbd344cbbe692a86c04707b4 : SUCCESS in 27m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/96681667efbb497d9977bfa693c836bb : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9dc3f85113ea03f179dce9a091954947e25a7718","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-24 16:16:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dae4832f7655f3a30a91973617e19e86f8c484b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:38:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/613c7595ed2f4e66acad8c631e377eb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2db2b66647146d99f0220eb2b641cba : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeb76ee954cf472c9a1c5bd14c9d073e : FAILURE in 10m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32751097428e4b898805f12c72d5ba25 : FAILURE in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9218f280784f09aa2cfe0d1f3876ac : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/852d461589b74c818fb683e62b8c2831 : FAILURE in 14m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c778d385af40457ebdc50037908cf722 : FAILURE in 14m 16s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5fa955c25c60466a8773ad9a2c7fa065 : FAILURE in 13m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24090c9dc5d84331ae62b3c04e3abd58 : FAILURE in 13m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69b752e3dc6c436ca6772422db0f714c : FAILURE in 13m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dccff3940b7e446e9f2a7154df4df562 : FAILURE in 12m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b1874db0e0874b1abbf5ea8125daf514 : FAILURE in 12m 33s","accounts_in_message":[],"_revision_number":2},{"id":"383b4a2cc1233d59d2e5e1e309ef7dbb1339acc3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-26 22:36:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a47abeb15f448933cdea8f53694aec81cee49371","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-26 22:37:03.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"ed8bdbdb96d2432a44000345c1342a222c54af93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-31 13:07:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f52490b1a57bffb26d16b13b731bc241c35543e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:56:16.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e36a1a4e88b14bfb9d459e406d206958\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e17d59b8bb340e5b346c33fb31a3ce7 : FAILURE in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/790877e11e0b4bee83077f773f60fa47 : SUCCESS in 13m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5aeee4461734f1589c57ccfd2d3abd6 : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afeca2d43d1b4c7d86e5309c1bd0da87 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f73eebc64dca4c39a366c73ac1fcd372 : SUCCESS in 12m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/24cbb068e1354b9d8515dea7ab9e036c : SUCCESS in 16m 01s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/9f3396e8d90e4dec9094c83de6d344d4 : SUCCESS in 34m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba18f20f45f54e279757f38a13ed5442 : SUCCESS in 24m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8257818bea0d4af98e0d55161a2d0eb6 : SUCCESS in 15m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4441304a156648daa5b809f5de1f6c94 : SUCCESS in 36m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1dc788fae8b04a6aad1388a76304d86f : SUCCESS in 25m 55s","accounts_in_message":[],"_revision_number":3},{"id":"fa0b509de163e2e8e0c8068db639c5288ea6e26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-31 14:43:25.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"30881ebcf6eaf9fbf310bf9bd2601b218d6b298d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 15:28:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3cb1c39efdb4cdfa8a50f62f4188f75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0c23e97f7a4c8fab88fcbe8446f1f8 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d405e81e44dd48eba1259a8b5d09c35f : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bca0b3ffc5db4cd7b706fcdeb863e9c6 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81803047d0654bc6a84f1dbf786391b1 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ff47afe2b7648799c6f0e4d4f311d39 : SUCCESS in 15m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/af9d4e7f14ee492a994ff303a0603bff : SUCCESS in 29m 35s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e6615c9d2535493eb3e360fec1e67222 : SUCCESS in 35m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/385ef8f5d7394914b7dafdb35e69a267 : SUCCESS in 15m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5640acb86b76429db0445168ca728eb6 : SUCCESS in 31m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e2d37ef1d2994d22841fdb8bc90c98f4 : SUCCESS in 32m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dd8c4ef8ccaa4a16afa621bc9079e4de : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e3342a3059d67a846eec07ee5737e207ad8f6e65","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-23 21:03:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"00501aad9c4f48305b9dea231992a136b2d792ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:04:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce4a424648fbdafb73a8100305fac34e8385fcc7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-23 21:04:16.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ea6a124f0f4f02289172c77927514bd21b93fdc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:59:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd228392d31c462bb610b8192026f41a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d93d97b30a24353a1d9710b58c4476a : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29fed0feb9db41e69b5df1280dd9ee55 : SUCCESS in 12m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/265babb1e9f34e6cbf2228a415a5d49f : SUCCESS in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8dbfc69cad445ab13a18e69112c2b2 : SUCCESS in 7m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7811f86ebe4e463aa29d2cfd00161e65 : SUCCESS in 34m 29s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/eee510b1d7a849f29e2af18100077f40 : SUCCESS in 35m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c035fa5f16f4a969b6bfee793fee144 : SUCCESS in 37m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d9561bcbd4f4332a27b673f872c6fed : SUCCESS in 34m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fec4fbad303c498f9bea182507eee574 : SUCCESS in 35m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/66afa50df78b4d7c97a7c31d30077104 : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":4},{"id":"44767808f71b9fd3e38d3f92c0f44228483c92c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:59:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b055ee8ebfdf496320d2648fb6b4f1abae10d42a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 22:01:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f17c32901d43ffa681b054e4bb505d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4777588a8814c479222761606aefbba : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"371a65fc875e82deef4215d2a98a707b67fa81b7","revisions":{"f26cf28bd8b9cac7852145e7cd12ddbe0917459b":{"kind":"REWORK","_number":1,"created":"2025-03-12 08:55:18.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/98/944098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/944098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/944098/1"}}},"commit":{"parents":[{"commit":"9fba42f3e0a9348e61430ab6769e2356df30e0d0","subject":"Add legacy action group to list_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9fba42f3e0a9348e61430ab6769e2356df30e0d0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"subject":"Support fnmatch glob for allow/deny lists","message":"Support fnmatch glob for allow/deny lists\n\nThis adds support to the config options introduced\nin [1] to use glob patterns matched using the fnmatch\nmodule.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015\n\nChange-Id: I2b874b7ab2d213a8859b7b4bbd9e546278595c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f26cf28bd8b9cac7852145e7cd12ddbe0917459b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f26cf28bd8b9cac7852145e7cd12ddbe0917459b"}]},"branch":"refs/heads/master"},"e1844a38e051c7dadb2b0c233944ce55c8030746":{"kind":"REWORK","_number":2,"created":"2025-03-12 08:59:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/98/944098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/944098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/944098/2"}}},"commit":{"parents":[{"commit":"9fba42f3e0a9348e61430ab6769e2356df30e0d0","subject":"Add legacy action group to list_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9fba42f3e0a9348e61430ab6769e2356df30e0d0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:58:54.000000000","tz":60},"subject":"Support fnmatch glob for allow/deny lists","message":"Support fnmatch glob for allow/deny lists\n\nThis adds support to the config options introduced\nin [1] to use glob patterns matched using the fnmatch\nmodule.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015\n\nChange-Id: I2b874b7ab2d213a8859b7b4bbd9e546278595c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e1844a38e051c7dadb2b0c233944ce55c8030746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e1844a38e051c7dadb2b0c233944ce55c8030746"}]},"branch":"refs/heads/master"},"a98efbb3c5fef86e142a5c14fe94788ea0b54c61":{"kind":"REWORK","_number":3,"created":"2025-03-31 13:07:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/98/944098/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/944098/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/944098/3"}}},"commit":{"parents":[{"commit":"9fba42f3e0a9348e61430ab6769e2356df30e0d0","subject":"Add legacy action group to list_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9fba42f3e0a9348e61430ab6769e2356df30e0d0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-31 13:07:02.000000000","tz":120},"subject":"Support fnmatch glob for allow/deny lists","message":"Support fnmatch glob for allow/deny lists\n\nThis adds support to the config options introduced\nin [1] to use glob patterns matched using the fnmatch\nmodule.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015\n\nChange-Id: I2b874b7ab2d213a8859b7b4bbd9e546278595c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a98efbb3c5fef86e142a5c14fe94788ea0b54c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a98efbb3c5fef86e142a5c14fe94788ea0b54c61"}]},"branch":"refs/heads/master"},"371a65fc875e82deef4215d2a98a707b67fa81b7":{"kind":"REWORK","_number":4,"created":"2025-03-31 14:43:25.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/98/944098/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/944098/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/944098/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/944098/4"}}},"commit":{"parents":[{"commit":"9fba42f3e0a9348e61430ab6769e2356df30e0d0","subject":"Add legacy action group to list_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9fba42f3e0a9348e61430ab6769e2356df30e0d0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-31 14:43:25.000000000","tz":0},"subject":"Support fnmatch glob for allow/deny lists","message":"Support fnmatch glob for allow/deny lists\n\nThis adds support to the config options introduced\nin [1] to use glob patterns matched using the fnmatch\nmodule.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015\n\nChange-Id: I2b874b7ab2d213a8859b7b4bbd9e546278595c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/371a65fc875e82deef4215d2a98a707b67fa81b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/371a65fc875e82deef4215d2a98a707b67fa81b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
