)]}'
{"id":"openstack%2Fmistral~944793","triplet_id":"openstack%2Fmistral~master~I5c8438b4d144faa0135933826c3eb55a722272b3","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-03-18 22:31:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I5c8438b4d144faa0135933826c3eb55a722272b3","subject":"Add Dockerfile","status":"MERGED","created":"2025-03-17 18:14:16.000000000","updated":"2025-03-18 23:38:25.000000000","submitted":"2025-03-18 23:37:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944793","meta_rev_id":"cf12bea6b98ebf7b80c77ccadd911c575bb9c73b","_number":944793,"virtual_id_number":944793,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-18 23:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-18 12:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-03-18 18:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-03-18 18:25:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-17 19:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-18 12:11:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e751caa8b308e34b7906f5ca6771863b8804bfcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-17 18:14:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96edfb607aa7af2e1356fd0b09e4429b428d96a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 19:06:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3195e6eb37b448fc96be1244b23a0f0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919d83000ddb474a8270e54b7b5e02a2 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4d22dfc69ab47b4b18e61e080d11fc1 : SUCCESS in 16m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/829b5f88b26045c4aef3aae50fb91d62 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e7c6cefba8469fb1c7d16880fc43c8 : SUCCESS in 8m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/730af7c390a84d058ec6764143c41de6 : FAILURE in 17m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a9bb7278f91d4e71a46f640ad15278b3 : SUCCESS in 46m 12s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/7fff986331a44f46898dacb67b31cde5 : FAILURE in 20m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6eaadfe221149ae9524210f0578b9cf : SUCCESS in 37m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dbc0a38e04046ffbfb74edd6c0cbdf6 : SUCCESS in 37m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f372e00511da4be3a587c42c2b461176 : SUCCESS in 45m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dbb6744366474577a32f2d4b3c094d4b : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":1},{"id":"66e6141424e260a234e1fa024dd4cbca1dd599bf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 12:11:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4bb95b6112ce432337addbde9f8a7556ca333f61","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 12:11:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79c13dfd7dca86f4fd9ca29f5b6f6f7f35ea05d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 12:59:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16a62a873144d20949b7704cb5fce3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb205b6aff2413695b0ae61bf26739c : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b92f16a7f4c447b79df1b3e380057ef3 : SUCCESS in 9m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71db4a4dae5a481086cb04a5ce744498 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1888c8e404c2445cb9aab2877e9ab370 : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dccc0eaae9c1415cb1953d4750756fa7 : SUCCESS in 15m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5d53932674624557a9db5580f679f075 : SUCCESS in 31m 31s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8d4f77bfbe9d47738e0d46f8843cbdbe : SUCCESS in 29m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/659aa39796294c2b9c1002b94e4d95c8 : SUCCESS in 32m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab3ff181bb6c407f838a7fe7a0fbe5dc : SUCCESS in 38m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/22b4855a419c4ace9d52825ac08c3fb0 : SUCCESS in 22m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2eb0de8aefc14791be1c6a23562ad6d8 : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9fcb85e4c08ce7279e951f45294910d967819f65","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-18 18:25:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1faac8dbbf9812b720028faf4f891dc4e92e7f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 18:25:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cb51c03da82b0472a64c4b39f1af8203aec0db1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 19:20:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70eb955ea84d4732afb6f7554cc7c3f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8eb04be437c4f59a09235d6f98687fd : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79c7a8a154704fa0a0c3c4f08fe3552b : SUCCESS in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d87aac4975040b9afdb2ecdfa43ff80 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f71430deb964198a4a7f79c2c7a458f : SUCCESS in 6m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fe596c1395fa4fda95a9f7261bfec3bc : SUCCESS in 21m 45s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5f06827095464961a5964abae314adc7 : SUCCESS in 33m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b67e57f352a42ea808c90df0f8be6c8 : SUCCESS in 32m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b9a6025fbd849bdae24990c9284278e : FAILURE in 41m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/486e8938e6e5430d90872ebf17ba94d7 : SUCCESS in 27m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/987a398e28be4d0aba7bf2b960f23047 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ce9472a65196d85b7ed27ccf6d10f5c1735e169b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 19:21:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8603746a3a9ee1ecb1310ad00e21d57a940ab91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 20:11:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a46b06166f8f49268cd38c6b23012ea4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec05401271f43539e3d5a30d0211afd : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b8c75138812418d8144d1de76bbd7aa : SUCCESS in 13m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00989fefabab4b6dbe0dc2a3d1945c3f : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c2c13ed2574fe7b6d10a03de4829e2 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b318048785ed42c4bc2f035342d16171 : FAILURE in 16m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c37f15d2240948f4b21af92f1de29374 : FAILURE in 38m 53s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/48e6ce99081f443d97799679f6908bdc : FAILURE in 33m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41dee8e9494d4826aa3fae1a3d2ce73e : SUCCESS in 37m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e061f975fab046fdb31ad93eda3bbe50 : FAILURE in 41m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7b753c0eebeb4969b8231ea98ad1c691 : SUCCESS in 35m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a118d9ca8d0c4069bb627106e61db002 : FAILURE in 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7356bc41bfe3841df64464de24ee75ff54304bd3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 21:02:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb12c5e60f64d3b0d093499fd5054c4402493aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 21:46:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7ba4edb11ce48c5a134ab16c1e140ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b7c7c4d2224de38789a1a75cf92ed4 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68ce5e099ec84a82874c3ee030a2ba29 : SUCCESS in 13m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9542c99eb220451e93a3f31d3d185758 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d721423ab14979b9540489a64d742e : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae5464056d264b8d9cc936b7bf771215 : FAILURE in 16m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7367f209298946148ed1601049d2a287 : SUCCESS in 38m 24s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/28a81dcfccf94f51ac555a043906bcc4 : SUCCESS in 27m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6be16efd24424198bbda8aea42e4e83b : FAILURE in 28m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ae066e2b72c4ca1916103f2cf678803 : SUCCESS in 26m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dd94bf9b0dc945c2a63eba763ad94f83 : FAILURE in 31m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d675476a023b4f40ac884b0983215a6d : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8a3b28917d6e7ac4b3d717e6a54b982415656fb8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 22:31:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f84f360e82dd82a558a0e75fa6f3198a3416f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:08:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72fbe8bdd4ec470090f910cacde297d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6665eede2d45dd931a8b1df25e6793 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ab39cd301f44eb7bea247d8c51deb35 : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/752456366df04a8ba15e286dc2045589 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec03253f8a24bc78d0c23bbf9fd5c71 : SUCCESS in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/638aaa4ddf6b412f8d4df2f80be8399d : FAILURE in 17m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/856876c9b6744fe29aeef869313c82f0 : SUCCESS in 35m 43s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/2549d6bfa6754a2aa42338cfb52733d5 : SUCCESS in 32m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccae47def6534c2e8576b5350128c88c : SUCCESS in 15m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4972988e5d6447008d41c715701fcd12 : SUCCESS in 31m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7b730129fdf34ec280efad46d5b5c853 : SUCCESS in 32m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7b666931f8984b42a170352b21efbc0f : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7bb3e44c7f4acd38784daadc3419431046a7ec3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"260fb03679af3df5dc32a47809a3e7d08b3de0c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:37:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3adaa6406994a088569a7b61bbb40de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145b7fefe1ed4dd2bb9f3e39001aaa53 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c1207e7a3db4f00aea53acf1214f46e : SUCCESS in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d62b4e2539349b5a3ba74529792d4cc : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d596b1d5c4fe43af99c922bac307f95e : SUCCESS in 8m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b31735b637d149708666ae670b0f45d5 : SUCCESS in 27m 30s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/e8ae5819fd3e4575a366eb173ce39cc6 : SUCCESS in 14m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bab1d2557644f3989d8f8ec8cab3c9c : SUCCESS in 28m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/130ad89f600942e19331d10c3b9839c7 : SUCCESS in 16m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5dbe0d19bdcf443ab1cab1950e446b30 : SUCCESS in 15m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e627088add3b4aef813919185b8cb1a0 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b21d54e141e26665619368fba774ebd23d280841","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:37:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf12bea6b98ebf7b80c77ccadd911c575bb9c73b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:38:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea07ce8aa7142078c6c0348e8e7c383\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/165583af81ad49d28b700bfefae00140 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"688c4e5743df5b9c2c81ac43fcdbdd817e25cf1d","revisions":{"688c4e5743df5b9c2c81ac43fcdbdd817e25cf1d":{"kind":"REWORK","_number":1,"created":"2025-03-17 18:14:16.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/93/944793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/944793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/944793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/944793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/944793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/944793/1"}}},"commit":{"parents":[{"commit":"fdb42356b45e809b7c4972fe5f25def8503829c2","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdb42356b45e809b7c4972fe5f25def8503829c2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-03-17 18:13:19.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-03-17 18:14:13.000000000","tz":60},"subject":"Add Dockerfile","message":"Add Dockerfile\n\nFor local builds, for testing / debug, it could be useful to have a\ndocker image.\nLet\u0027s introduce that here\n\nChange-Id: I5c8438b4d144faa0135933826c3eb55a722272b3\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/688c4e5743df5b9c2c81ac43fcdbdd817e25cf1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/688c4e5743df5b9c2c81ac43fcdbdd817e25cf1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
