)]}'
{"id":"openstack%2Fmistral~948013","triplet_id":"openstack%2Fmistral~master~I6b00bc6538e056b30fe4aaac6494a1a38781fb86","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I6b00bc6538e056b30fe4aaac6494a1a38781fb86","subject":"Force yaql 3.0.0","status":"MERGED","created":"2025-04-23 21:29:20.000000000","updated":"2025-04-28 19:38:30.000000000","submitted":"2025-04-25 18:56:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948013","meta_rev_id":"ee14f2114463cd93826f4035b624ea7ac4b2ffd5","_number":948013,"virtual_id_number":948013,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-25 18:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-28 19:38:30.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-25 18:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-25 18:15:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-23 22:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 19:38:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"47baa962120da6960532eceaf6dac05f77c240e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-23 21:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18b1374d83f4c661d759950d7da706f778e83b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 22:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6730fda4b9fb4ff2888320ab58ba2cd4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3cbfa1eb1d4023897c9478634b8528 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23d72e39e6e84555b96b08c5441d8264 : SUCCESS in 11m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd91b1d543f247c1b5924cbf82b86750 : SUCCESS in 11m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9caa4072d2d141e29bcb84ffd4e3bf14 : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019a341d8f364e55ad867de105c192ac : SUCCESS in 8m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/28b684bcd1054a708d37fbd8bb796a46 : SUCCESS in 3m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a891eb856fa485888534b08d74cdbda : FAILURE in 15m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fb0b94a01c3b4bd690eb7455911e045c : SUCCESS in 26m 37s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/4037e035ff0a41029496870e26439794 : SUCCESS in 38m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c815605c1830449b86ce9ecbc8fe5ec5 : SUCCESS in 39m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f9206c314649a9800d341d117ed2bd : SUCCESS in 27m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/72ef854569b64e40a43a81f83b8e1a56 : SUCCESS in 27m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2a6a080e58e245b7ad5e5498af70d569 : SUCCESS in 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4bf1a7f51cfe9fd58f4e7cc45652a056f8aef7eb","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-25 18:15:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf8bf4c13705034b61d38528049ff3518a4a0a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:15:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08bf25c47becc6def641c92a050e16c166a21d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:56:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6efc8dd172a94215832133ec2d4c2770\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d080c3cd4944bee927df3189d76dd27 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfa30cda251f4ddfa110e0873106d918 : SUCCESS in 9m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06c053b1780248aabd58068cbfd01123 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dda173308e2439b9f6ee3eddc21954c : SUCCESS in 7m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6a5a51847eb492c902989dfce90700e : SUCCESS in 3m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/000bb535f29546c9b67037b0b26ea7a6 : SUCCESS in 31m 01s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/16461ce3af8542148ed381854b5a58df : SUCCESS in 33m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00232c79dfe346f98696aab695b70e61 : SUCCESS in 30m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e083fbf082b441292b80dcddeb54c71 : SUCCESS in 16m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d992b694b566431f9c5c45371561fad8 : SUCCESS in 39m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c941b683417c4f679b3a4cfbe773e38d : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":1},{"id":"58ce0df696e93dc4146a15e44935a7d058813583","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:56:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08d42112cd73aef6e93e05bff3c651c2550032b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:56:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b745973359e4d688b7ba41e87d23d5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f79303fc4c42e88597c9192e90da8a : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1},{"id":"ee14f2114463cd93826f4035b624ea7ac4b2ffd5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 19:38:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"025d3e5ed771db5ecc23fe3115de6189369339cb","revisions":{"025d3e5ed771db5ecc23fe3115de6189369339cb":{"kind":"REWORK","_number":1,"created":"2025-04-23 21:29:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/13/948013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/13/948013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/13/948013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/13/948013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/13/948013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/13/948013/1"}}},"commit":{"parents":[{"commit":"688c4e5743df5b9c2c81ac43fcdbdd817e25cf1d","subject":"Add Dockerfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/688c4e5743df5b9c2c81ac43fcdbdd817e25cf1d"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:28:06.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:29:15.000000000","tz":120},"subject":"Force yaql 3.0.0","message":"Force yaql 3.0.0\n\nSince yaql 3.0.0 has landed last year, we don\u0027t want our users to use a\ndeprecated version of yaql in mistral.\n\nLet\u0027s force this upgrade\n\nChange-Id: I6b00bc6538e056b30fe4aaac6494a1a38781fb86\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/025d3e5ed771db5ecc23fe3115de6189369339cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/025d3e5ed771db5ecc23fe3115de6189369339cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
