)]}'
{"id":"openstack%2Fmistral~948018","triplet_id":"openstack%2Fmistral~master~I0164e2a38c174156510d65f1c326f6fb646687bd","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-04-26 12:16:14.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-28 19:40:22.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I0164e2a38c174156510d65f1c326f6fb646687bd","subject":"Get rid of coordination","status":"MERGED","created":"2025-04-23 21:45:24.000000000","updated":"2025-04-28 20:53:50.000000000","submitted":"2025-04-28 20:52:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"948018","meta_rev_id":"49067df487c7c2add8320dbd5fb39a2f520b2c04","_number":948018,"virtual_id_number":948018,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-28 20:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-04-28 14:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-28 19:38:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-28 19:40:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-28 19:40:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-23 22:06:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-04-23 22:31:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 14:37:56.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2025-04-28 19:38:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"8b91adeab53b564b86c41b2866f6c855f223cd99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-23 21:45:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0157e9ea98eeb2df4f43ea45476640a1c32590b1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-23 21:46:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c81d4e40567ebba09d4b1ab45fa8ce45a3f231d2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-23 22:06:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"813cd2dde807b7babb8f50d270fe6fbb2a055358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 22:31:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba29819f5a48434c9091df83ab44aa24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9ea917d9344b05a98028266f072120 : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbdbde557eac437eb5a556fbefe3050f : FAILURE in 9m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a3cd85a1ff94defafab896a5eb54dd9 : FAILURE in 9m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8032fe823a554f9ebf0ca9c297f69269 : FAILURE in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddec8056e1248f58188f8ef481dae54 : FAILURE in 3m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/addc9a08746e4ba0bd552e75ff71a49a : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafeffe2c3b44f07adf2e50621d04082 : FAILURE in 16m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3ca08a10af984bff976ce0e5693ec984 : FAILURE in 40m 43s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/6dd5baf98bf9426a965a092f0ba28bd2 : FAILURE in 13m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9beb9ffde4564ed4b118dfe9fe9f56f9 : FAILURE in 28m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a46ead90f95d4411b6cacd49b4a8cbbf : FAILURE in 21m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/110111e184ae489da95cc853041b09fc : FAILURE in 19m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ef0e6b97164e4cd1bb58b522378eab5f : FAILURE in 19m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a45a68fdce0958d23366af75b83214f6f363bf0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-25 18:14:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d51422dc351a002fd20305935bcbe8988bfabb37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 19:10:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d229c74ce5d4bda907ca8656b89b9f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c6a4c3a78745ec89033027511eeace : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a461fc64a68d40dbb84268b2cca288e2 : SUCCESS in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/348242f0bc0048af8f83e2f7dd0c167b : SUCCESS in 11m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf265cd3ac074d3f9af275edd548f6a0 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803ee5c7ce784598af18b5376581fc5d : FAILURE in 7m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da6002c6d8e845a189583fa1669a5d9b : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7f855bee9d4494fa771c69e81598267 : SUCCESS in 13m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/89445d9b265849f8824ba860cc47f9a7 : SUCCESS in 42m 47s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/45419e154d5a40a9afdc7b52b12e8a73 : SUCCESS in 14m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b3a1bba4afe4bcebec69b36c453496f : SUCCESS in 50m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6235a55de7de48aea2cb6763ad32aba3 : SUCCESS in 36m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3faa6c4ccc07414cb18c5fbfe122dda0 : SUCCESS in 35m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1e7534d2e8bf4e6487ea20dcc0698aab : SUCCESS in 26m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9239880cf4f2988efe7431677bc2977e5a3efa7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-26 07:02:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"188bbbafe3dd670cb921649e1d3b8db56c61d801","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-26 07:02:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd0b1bef412cb6db7e5166f90f467a04c09b52ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-26 07:08:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7c6c79889ef651284d01171b54f9e402706f9272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 07:55:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a205ec81894846829d3f327e6ac4f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f8b278e1fb436480006404b0b129dd : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da51a5f7e6624a869999b16a8c028b3b : SUCCESS in 9m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b3d4fb1099648478cbb83185b3381d1 : SUCCESS in 8m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f5f42b47d1a4b6684eb12d22db3a776 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5d0986397b48d49a7be1f6b8444249 : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a63b9d9a741f43fb8e84552249a14e59 : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5512f37767444238b53890542bdb887 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/176e708f0f2d496697c414aa0bf4a49d : SUCCESS in 14m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ce542c7678fa4a3b983e0783e7e8338c : SUCCESS in 20m 15s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8d460a4828e241188f701679e9564d1c : SUCCESS in 19m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d190c76461f4d98bf623be80355d724 : SUCCESS in 19m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ab8f8b1b39245989fc2a320eef6b3d9 : SUCCESS in 20m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/578119fa0ca44ad0b76ab7f214ca1096 : SUCCESS in 36m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e0b5a6cb2ccc4bb4873bfa004fe84f64 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":4},{"id":"a90f01b38a8d88f454b9d3a5f1a3849d1772d1a4","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-04-28 14:37:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3b9d4e4aa9f2acb27f0c8d79e3acf707e1bbc2ed","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 19:38:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"440a2e4837a3241fe12bd354bf94c776f8684753","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 19:40:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6d2bacb8d0449b26f9e3efac1f059578842924ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 19:40:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23f624bb038af543db290d3df1af45df612f05be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:52:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd82af46a8444f98548d378674e3e34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d89dc5550bf46d5811b9826519cd972 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82b80f598ede4c88b7cff4020dad3604 : SUCCESS in 13m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e773ec48a66948778d6a75b313dde280 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41911140cc484b4bb60cc1b6307e311e : SUCCESS in 9m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b76ddf16e23d47d6b7b3292cc570f108 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8e8e962a30c4328883030920da10ae7 : SUCCESS in 2m 50s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/59f33ceb7df84655892b56d250b2710d : SUCCESS in 16m 29s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/64379f44bbcd4abea2bf432a21126105 : SUCCESS in 36m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c76d58a8a318477f966cf7e898858290 : SUCCESS in 51m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c4449ac1cc94f5582d5919b34ddaf27 : SUCCESS in 40m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f06dcc2aef30427ea9da1ade7a825d8a : SUCCESS in 51m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fefbacae7ba9461ea630b80e4067dd6f : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b8e952aad94e0f47f240eb938cbd0d4a1c1a07d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:52:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"49067df487c7c2add8320dbd5fb39a2f520b2c04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:53:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309e461745064c9a85a40978be01f6de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0f45d28cb741d2a7b4bfd540e38160 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/835a8b9c1f114dda8ffd13153a883047 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6a2e451e21032ae983729bbf4ade0c5d82c0929d","revisions":{"a5978cdc1ba8b02865b1bd0584425884d688b72f":{"kind":"REWORK","_number":1,"created":"2025-04-23 21:45:24.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/948018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/948018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/948018/1"}}},"commit":{"parents":[{"commit":"025d3e5ed771db5ecc23fe3115de6189369339cb","subject":"Force yaql 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/025d3e5ed771db5ecc23fe3115de6189369339cb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:44:02.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:44:02.000000000","tz":120},"subject":"Get rid of coordination","message":"Get rid of coordination\n\nWe deprecated this feature in the previous release of mistral, let\u0027s get\nrid of this now.\n\nChange-Id: I0164e2a38c174156510d65f1c326f6fb646687bd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a5978cdc1ba8b02865b1bd0584425884d688b72f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a5978cdc1ba8b02865b1bd0584425884d688b72f"}]},"branch":"refs/heads/master"},"28fe20a76712787cbc4b871f575861cde6df59ad":{"kind":"REWORK","_number":2,"created":"2025-04-25 18:14:51.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/948018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/948018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/948018/2"}}},"commit":{"parents":[{"commit":"025d3e5ed771db5ecc23fe3115de6189369339cb","subject":"Force yaql 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/025d3e5ed771db5ecc23fe3115de6189369339cb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:44:02.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-25 18:14:43.000000000","tz":120},"subject":"Get rid of coordination","message":"Get rid of coordination\n\nWe deprecated this feature in the previous release of mistral, let\u0027s get\nrid of this now.\n\nChange-Id: I0164e2a38c174156510d65f1c326f6fb646687bd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28fe20a76712787cbc4b871f575861cde6df59ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28fe20a76712787cbc4b871f575861cde6df59ad"}]},"branch":"refs/heads/master"},"73d2f992f23120138fd18a0b57cb17aaa4b3d059":{"kind":"REWORK","_number":3,"created":"2025-04-26 07:02:29.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/948018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/948018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/948018/3"}}},"commit":{"parents":[{"commit":"025d3e5ed771db5ecc23fe3115de6189369339cb","subject":"Force yaql 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/025d3e5ed771db5ecc23fe3115de6189369339cb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:44:02.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-26 07:02:23.000000000","tz":120},"subject":"Get rid of coordination","message":"Get rid of coordination\n\nWe deprecated this feature in the previous release of mistral, let\u0027s get\nrid of this now.\n\nChange-Id: I0164e2a38c174156510d65f1c326f6fb646687bd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/73d2f992f23120138fd18a0b57cb17aaa4b3d059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/73d2f992f23120138fd18a0b57cb17aaa4b3d059"}]},"branch":"refs/heads/master"},"6a2e451e21032ae983729bbf4ade0c5d82c0929d":{"kind":"REWORK","_number":4,"created":"2025-04-26 07:08:08.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/948018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/948018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/948018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/948018/4"}}},"commit":{"parents":[{"commit":"73221b74c42a6ee509cce56c2a2ca6649591d352","subject":"Merge \"Force yaql 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/73221b74c42a6ee509cce56c2a2ca6649591d352"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-23 21:44:02.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-26 07:07:53.000000000","tz":120},"subject":"Get rid of coordination","message":"Get rid of coordination\n\nWe deprecated this feature in the previous release of mistral, let\u0027s get\nrid of this now.\n\nChange-Id: I0164e2a38c174156510d65f1c326f6fb646687bd\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a2e451e21032ae983729bbf4ade0c5d82c0929d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a2e451e21032ae983729bbf4ade0c5d82c0929d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
