)]}'
{"id":"openstack%2Fmistral~948181","triplet_id":"openstack%2Fmistral~stable%2F2025.1~I57cd468cba227cc8cc7960e84e9a111217e80cad","project":"openstack/mistral","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-04-25 21:10:36.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-25 10:41:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I57cd468cba227cc8cc7960e84e9a111217e80cad","subject":"Add allowlist and denylist for action providers and actions","status":"MERGED","created":"2025-04-25 08:24:12.000000000","updated":"2025-04-25 21:11:19.000000000","submitted":"2025-04-25 21:10:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948181","meta_rev_id":"252a698331a8e073c9032de3b119f90e3a956e4c","_number":948181,"virtual_id_number":948181,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-25 21:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-25 09:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-25 10:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-25 10:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 09:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-25 09:08:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-25 09:08:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"948049395413d9021c4fba4ac28ef00e964cce5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-25 08:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92e0e579526100632c117fdced20eb566116edfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 09:05:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9377f117d2ed40d089a1a659acf1ee94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3ea535c0d541ce90af4c9e98c6a8bf : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08cb4559ebbd4b248d731106f05a1163 : SUCCESS in 12m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7514490f57134e638c9434a965928c72 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e217b2c08b134e5b8a7ca4dd7acc7a6e : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ce351b132614067b14f10ad10847a8b : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42817248d4fc4b58a58a1dca49ebb6bc : FAILURE in 16m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3a22f21dbf7c4c80940bdbdcd89b071d : SUCCESS in 34m 32s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/a7b45aecd9f04a0da000c9e3d64361a5 : SUCCESS in 34m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59117515326b4528b250d1ce0bcebd28 : SUCCESS in 27m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5aa2489f590b4f7593e2c180578c2b2b : SUCCESS in 35m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8aa163cefed0471089c04c4fc75e7187 : SUCCESS in 36m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/189e5c7acfd94578af29c61817fc31fc : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b51c0e8db6396bfde6dfc9370e88a3ea01e344a4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-25 09:08:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4a6ed52284ca4efdd22f536d681a8ac23361216e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-25 10:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e254652d5100678ddafefb9aa96f6ecec37cd4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 10:42:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c6be8b79fa8bc1d369ccaf2b276ad5d71ab0b6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 11:24:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0affd506deab4b79afb8a8b19057dca1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac323c77bb0a4438b92873a898d45c04 : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/585716f5a26f48e3a7e42d0156cb9971 : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1472f83b43744ea0bbebfa7ba4449863 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b672c2568a3e4d159ba0ec68e902cf11 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d83ed25b1148ffbc22c75f8d4f40f7 : SUCCESS in 2m 45s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0a528d66c69140298c6aaf8409e6fc21 : SUCCESS in 37m 42s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8da4e3213b0c4ee28c41a831831270da : SUCCESS in 33m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/594feddcf1174d25bba97a11788ecea2 : SUCCESS in 17m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33dc415ea4d94fa1a43a40052a6d2489 : SUCCESS in 35m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f76c54f8c2c041699a52b56e18614f14 : SUCCESS in 18m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/869dbeb5e6ff4b089aa6c0fb1108ed3f : FAILURE in 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7f33a6ddb2da1fdf9548d91194c7d491ee02aa9d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-25 11:52:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbcc2e9a9c02a8ef935827b22db41d8160c224db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 12:52:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d2e4182d18f4e268a51950c023042e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9078ea303c8345ddb26ea44fd566e51e : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e189f34feb96447dbe98fdd8bdf7445e : SUCCESS in 10m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bb71e9549f84bba8de34f462b9ef215 : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36212883ad33426f90581fc7fbffa40d : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0627ceda13f1452e8325c4f70ca1ad6b : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e32e589aee34fa5840f99a9e405792b : SUCCESS in 15m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c8eb4b8df2494d23995f6415a63d2d61 : SUCCESS in 17m 28s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/4e27e8d65dbb4aa2ad525e6e45478a88 : SUCCESS in 32m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf644fdcb2c94aa29ee34cfb72a8ae7d : SUCCESS in 36m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30234eccce7648eb8ac49f36bc653b17 : SUCCESS in 26m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1024e4f9ad9c4230a7774c367e2aa6f6 : SUCCESS in 35m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/09566781383e4fca937f45e38598633d : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"580ed968ba5f70ea8834b9e06ee9d3f7ce9d2610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 12:53:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1d5be68cc72207b121f1220f26fc500ed24fbc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 13:37:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c56a737b1a84cf5a8fcfd8bca1d1407\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf51e47b2c244297beb92bbb1f57b9d4 : SUCCESS in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbeaf2f2e970422eaa76ee160dcc4b6e : SUCCESS in 12m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e869901943304e6fb56ed0a92c90efca : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dc7629661c49828e290c3813a4e1d3 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46ad50a39d0c4c7491f3b3cc6493d6a1 : SUCCESS in 2m 52s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cfd51c2a83614c4999a55dff42dcbba2 : FAILURE in 21m 40s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/5dd0d3c9825848ebb9e8e4e15db01d76 : FAILURE in 35m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8ca3c4809094e3989fb1493cd6743a6 : SUCCESS in 37m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04ccf601f7814e1e8a9e18d5a2f83a63 : FAILURE in 25m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ac94138c14ad4d1ba4d9baf0e5793501 : FAILURE in 40m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/76713e810e1c4f96a741fc0cfa0f1dcd : FAILURE in 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0e6c1586e61fd8c573ff39a888a14df51c9bacff","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-25 13:53:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb4a1c2b76e6ecaf6374bc272bfd8e4c897ab99c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 14:35:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cb1875382334166aadc7ba7147f333e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c72308676834101b00f937fcfe4ad17 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eeeb1dd1aa34b4fbb69a0cdd7cf5b2f : SUCCESS in 14m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ae6d2edeb5447e3bb9050a4fd0028a8 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0305fed8797443828faa8eef6d6aa375 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42d2e0d532dd4b7b9a99b46dee10ae80 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ab9736d8dbb48bf8a6ded56e181c941 : FAILURE in 21m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/93ed0477c21d4fc988f362e92f00da20 : FAILURE in 33m 14s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/cabecb7bd685432c80f0689225eb5fd7 : SUCCESS in 33m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f49fbc7fb5474492b62ba3faaaa9c93d : SUCCESS in 37m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b53433cc2af4573971ed77a6a80810f : SUCCESS in 32m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/67b7b5633d6046dd9083fc44f86c7699 : SUCCESS in 31m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2c294d02b0474e008c3cdfaede376f58 : FAILURE in 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4a8041c919c2aa0c0a5b65edd267a07bcf100350","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-25 18:28:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05957ad7ca66c3155243904fb1883a3dc0ec17c1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-25 18:28:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57d41b234f976fee2bab807499fb8bf18835f33d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 19:10:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ff19f2b9fef4044a43f8419e3295814\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b60bcf02c7411683d68f4ffd86eeff : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e411863232d44dcca077cf08bd97cbcc : SUCCESS in 9m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/935a584ba6654d2d894fcf0a1a7948c4 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5245baabf6e3423aad9dd39b9f83ab8a : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f42e0b9cc499420faf2e9f577e7ae778 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71f4b0d1799946529c8eddc08eed62de : SUCCESS in 12m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dc841fe33090491681ffad8fddf8e27d : SUCCESS in 15m 44s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/1e0732fd465b419ead8398da2192cdd9 : SUCCESS in 37m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52d0a34b31f040cc9c084062624438f8 : FAILURE in 16m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1fcac1189814a4f8a33da3042a20c61 : SUCCESS in 31m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b0125d87a9134e56822f9ee630c9a462 : SUCCESS in 29m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9b8de2a693c34384a9a49b27eef14abf : SUCCESS in 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3b1290ca153d69d0df7c4c6a7be91987a94eacde","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-25 19:41:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9accab4e0c535df96c09914871ca18360d855bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 20:25:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1598868b28404f5b927b834ac4b5acaf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba2f17702658422988ce3f6e16e32c85 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adb308d2d0484260a9fc4e326069bee8 : SUCCESS in 12m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66bc971e02744026b359d7c001ea307d : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5f773d0d464cd5a6507ffc2eaae31e : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17ec7778dc8d4a8ba18dcdec66c253aa : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46e92a5f8adf48348a8fb230ebf41e1f : SUCCESS in 12m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/00ac01f1138e4fbc82caa2f756a5a40d : SUCCESS in 24m 21s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/96179e209d9649e7a998accda4f4f0d1 : SUCCESS in 35m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b026751d014460e960b4d5168e9af3c : SUCCESS in 36m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a4af861a591486ab9a7517d52ebd3f0 : SUCCESS in 38m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a8353fd522aa42baa2a73d51a155225f : SUCCESS in 38m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c5f9aed00fcb4992b5b3fc1691a73522 : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"808baeb44d7f5c84006dca5ec90c61598840bca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 20:25:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c173f7e858c5968fc2d234df203e01e6661af94c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 21:10:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5275fd0b842f48ae8b3150d98374e622\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ba58fe0bf74ad7b1b83939a7497953 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c649f8d26c24de5862a8975e52395f8 : SUCCESS in 13m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bb4580e96a14d199c8a13bcc65c4ef1 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fbb8f0294f4399bd05de544599dc8a : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af5857378d674ba9b0129387b282f276 : SUCCESS in 1m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b18cb30d8ac3486986bb0de6c9465646 : SUCCESS in 39m 56s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/b76db92568c2408799d76537e6363e77 : SUCCESS in 26m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3718a467148f409a9a8b26453889a05e : SUCCESS in 24m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41717f4c98ab4e3b9e2e74014cc05c95 : SUCCESS in 15m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/de7321966da3479ab1c0f8e7f621884f : SUCCESS in 35m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ee13d56c84294389a49f0f6cf3cb8111 : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"94ddfde43e98367d4782b25f13d5150b0785b2c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 21:10:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"252a698331a8e073c9032de3b119f90e3a956e4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 21:11:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c381be774542109ec2f4c1dc7d157e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b07937117341008436746980f111f7 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d7e7f20161b47fdbcb8035ab2a621fc : SUCCESS in 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e5d9c6ba1cb4a22363b04a548c078600753ae42","revisions":{"9e5d9c6ba1cb4a22363b04a548c078600753ae42":{"kind":"REWORK","_number":1,"created":"2025-04-25 08:24:12.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/81/948181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/948181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/948181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/948181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/948181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/948181/1"}}},"commit":{"parents":[{"commit":"40c63aa89ef3a3121d934e77e7bf7aadf52e6e7f","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/40c63aa89ef3a3121d934e77e7bf7aadf52e6e7f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-25 08:23:47.000000000","tz":120},"subject":"Add allowlist and denylist for action providers and actions","message":"Add allowlist and denylist for action providers and actions\n\nThis adds allowlist and denylist ListOpt config options for\nfiltering what action providers is loaded\n\nAlso adds allowlist and denylist ListOpt config options for\nfiltering what actions is loaded by the legacy action provider.\n\nThis is used in a security context where we need to restrict\na environment to only load actions using the legacy action\nprovider and then filter what actions is explicitly allowed\nor not allowed to be used.\n\nChange-Id: I57cd468cba227cc8cc7960e84e9a111217e80cad\n(cherry picked from commit 11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9e5d9c6ba1cb4a22363b04a548c078600753ae42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9e5d9c6ba1cb4a22363b04a548c078600753ae42"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
