)]}'
{"id":"openstack%2Fmistral~948307","triplet_id":"openstack%2Fmistral~stable%2F2024.2~I57cd468cba227cc8cc7960e84e9a111217e80cad","project":"openstack/mistral","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-04-28 14:38:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57cd468cba227cc8cc7960e84e9a111217e80cad","subject":"Add allowlist and denylist for action providers and actions","status":"MERGED","created":"2025-04-28 06:04:36.000000000","updated":"2025-04-28 14:39:39.000000000","submitted":"2025-04-28 14:38:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948307","meta_rev_id":"11c9afa201061b30cfc4b2d0c8740a4dbc0877e6","_number":948307,"virtual_id_number":948307,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-28 14:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-28 11:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-28 06:57:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-28 13:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-28 06:57:57.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-04-28 08:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 11:59:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"fa3fb924471a172343517fda46b27ebab4060f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-28 06:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac31e2682c5f45ede1695ffc1c1eeed3a4dc9460","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 06:57:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56d4a3d20d49e8b8cbc727386e376dd0d54500d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 08:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d52119d22da347528cc94cf26f2b376e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762dfa445f2c404b8566caca43dbebca : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb356fd0a3f743049424ab9546e55458 : SUCCESS in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc1be96a956045efb84e079a3aa4cb90 : SUCCESS in 11m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c3a924a2ccb487ca141b8fc9d5eb08a : FAILURE in 11m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7748bc17771349069f0b7ec3fa0cc722 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c05b558d7784993bcb2589eca9309c6 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75cf18eb8590424fb5cd4f25240ae6f7 : SUCCESS in 14m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7bbf5a8f304a48bf8677da9476127a0f : SUCCESS in 15m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f305cdb21d1d497d9ef653d4a21c8ebc : SUCCESS in 32m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5a23b89b72d4e2182cbad7cc5289c43 : SUCCESS in 23m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/724e3d6780ca40f8ab35896bad96e6d1 : SUCCESS in 34m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/875951c2b7a04041b05f2efa8ca6b38e : SUCCESS in 33m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d285984a4b5e4431bdcbe39ae0fdfbc4 : SUCCESS in 53m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fa257acb3aee48a0800fe073d17d006c : SUCCESS in 41m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/70cbdcf51890457cb209c5d3a3ecff92 : TIMED_OUT in 2h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c484e1c2e7d54339d9709dc5464564eeaa4e602","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 11:59:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dfeb65c1962c5a3f371096093e076e454f92546e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 13:36:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a40875955490b78c5414f32b993477eeaee63ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 13:37:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f5800d2537482d3441aa4c97092846835120072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 14:38:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/913734f94b09462da9cc131144322510\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc42ec9eb184b17b6eaba685f36087e : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/392edefb490f4bb787209d8d2db1fde4 : SUCCESS in 12m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/739a1720202943efab3869c86a063cbd : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/551892b3e59e4468a41f98d189dbf0c6 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ca69f5a1cf4a3483fd5b584578a19e : SUCCESS in 3m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/af11c0894c5940ae81e621b0ab9f3afa : SUCCESS in 34m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32d93b90c58f4e3490e75e288d00d727 : SUCCESS in 33m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3d65b2a9a6e41ff9777e1cbe624e4a5 : SUCCESS in 25m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d9c9972efeac41fdb3a5e5eb1f174462 : SUCCESS in 36m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fc817f0f732d4409a11ebeadaf565c02 : SUCCESS in 32m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ed6c8765a7334d288ee459564443296b : SUCCESS in 53m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/079a5709f6b84a58a4c05cba3665e1bb : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ddfc7c07f8ae9652e0c9fabecf0daba827f95f86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 14:38:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11c9afa201061b30cfc4b2d0c8740a4dbc0877e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 14:39:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f6ccecbc434a4aa17a4ea7ec7e36b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc116f7415540618eabd85769e73a49 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e1222b7edc3490a99617e65dfa2b7a0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a50e05ebfceb0129aea4ca684e0ddb709393c8a","revisions":{"9a50e05ebfceb0129aea4ca684e0ddb709393c8a":{"kind":"REWORK","_number":1,"created":"2025-04-28 06:04:36.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/07/948307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/07/948307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/948307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/948307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/948307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/07/948307/1"}}},"commit":{"parents":[{"commit":"c4d32f8ae4c56955a97031f087b8e269ec8e45ab","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4d32f8ae4c56955a97031f087b8e269ec8e45ab"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-28 06:00:45.000000000","tz":120},"subject":"Add allowlist and denylist for action providers and actions","message":"Add allowlist and denylist for action providers and actions\n\nThis adds allowlist and denylist ListOpt config options for\nfiltering what action providers is loaded\n\nAlso adds allowlist and denylist ListOpt config options for\nfiltering what actions is loaded by the legacy action provider.\n\nThis is used in a security context where we need to restrict\na environment to only load actions using the legacy action\nprovider and then filter what actions is explicitly allowed\nor not allowed to be used.\n\nChange-Id: I57cd468cba227cc8cc7960e84e9a111217e80cad\n(cherry picked from commit 11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e)\n(cherry picked from commit 9e5d9c6ba1cb4a22363b04a548c078600753ae42)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a50e05ebfceb0129aea4ca684e0ddb709393c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a50e05ebfceb0129aea4ca684e0ddb709393c8a"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
