)]}'
{"id":"openstack%2Fmistral~948308","triplet_id":"openstack%2Fmistral~stable%2F2024.2~Ic536029e30ebb2f620e4c3f99a53a7d400fbb9cd","project":"openstack/mistral","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2025-05-02 16:16:43.000000000","reason":"\u003cGERRIT_ACCOUNT_35600\u003e replied on the change","reason_account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-07 08:13:44.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-05-07 06:08:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ic536029e30ebb2f620e4c3f99a53a7d400fbb9cd","subject":"Add legacy action group to list_opts","status":"MERGED","created":"2025-04-28 06:04:36.000000000","updated":"2025-05-07 08:14:38.000000000","submitted":"2025-05-07 08:13:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948308","meta_rev_id":"0e5bc5c9201b7a08f104e46eb54882c122dd84da","_number":948308,"virtual_id_number":948308,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 08:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-02 16:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-05-07 06:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":1,"date":"2025-05-02 16:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-28 06:58:05.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-04-28 08:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 11:59:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-02 16:08:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"7bdcad5f726c70151be9264bf8a4f79edd0134d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-28 06:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66bea62921aba4fba2a4a337c99a342d19204b14","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 06:58:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7dd97eee521c7272fa93870566220b5c85215fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 08:08:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613fff864c9141a9b408332055091b64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df25de74c34b4aefa6d67d8840a2c6f7 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6715c413f46546518c23bc2cee5cd83f : SUCCESS in 12m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc3447ee04194f61833769eef1b1f408 : SUCCESS in 12m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48f7a3923fef49dabf9b93af94991b21 : FAILURE in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf9d512ab174b6898af377faee706ae : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd1685810f034c1fba5877fcdc6a6e3d : SUCCESS in 12m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/be15452ef02a41e5ac129f1e3489b226 : SUCCESS in 25m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39f92b5a8b864bad95b05dbdf6a9ae95 : SUCCESS in 36m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/293bf2ba68e444838209931c6af70d9c : SUCCESS in 26m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/143a7a4bf74e4f3cac325ca9b83bd56e : SUCCESS in 24m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c32dfc28190b4bf2b9634aa6d9992874 : SUCCESS in 34m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c7de8a8661064c8da0136569d2689d3a : SUCCESS in 43m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/95d53f72f2734fcdbb913cb56b0c6cf2 : SUCCESS in 52m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/434b0c1471fa40e2b80d0fe5709327e4 : TIMED_OUT in 2h 02m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59c759475f0e4d223c3e933d34d3a90f48efebf1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 11:59:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4180fa98abec47f9c1c9006326f28bd5226a6b8d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 13:36:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35ddbb3da3fc545e21d3f2aab667d6be91035e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 13:37:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d798e4383c2b9e206d01ecf6df0d853225883b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 14:40:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbd2a54ba2c14bf6a6d88ee9d3c7ca82\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99534773db54f7c85637f6356e86cf1 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f1f84a1c01844228452a7f4f4969e88 : SUCCESS in 13m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb94d8e139f24804a9b89bef936ed8f5 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669cd742c4ae4bcab24d640aedb85aa2 : SUCCESS in 6m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e10f55bb3d2e4599b01ba11cc72fd506 : SUCCESS in 35m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/048b0e4b9e66401fa6bde9b2fe5025c7 : FAILURE in 37m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16471200dd5d4056ad1a07e3eaf5f841 : SUCCESS in 34m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2898cc8c451445279683e016f0c0edcc : FAILURE in 24m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/10c180399de44723914ca6b4fc0e170c : SUCCESS in 26m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/16efa6461fb14fad86dec06a9d1cb1e2 : SUCCESS in 53m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bf8a58e1c03648e69660002925a738a6 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":1},{"id":"94ba371fd15a53a5b221d079b1386eee15727cf3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 14:41:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"638e51849b18ef2ff8a3f625f3085dcce5c350da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 16:52:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b71d99fd8074895ae13ba159c5b6d26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc1d6e069cb4719a9a13644435541e1 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c34bcf20261434ea8612aac6f63cfd3 : SUCCESS in 13m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d8a5944f21e4232918587bed670675a : SUCCESS in 13m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2376e76133c4bc6b76e3bb5a40f748e : FAILURE in 11m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c32a8135a8fd4096ae4063c765a58b29 : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd248ae58c374a8bb2abe743b976f009 : SUCCESS in 15m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6f802cdd26584d2e9c689cbc886c373d : FAILURE in 26m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b1f56ca412a45f6bd1de0154c62de00 : SUCCESS in 35m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef6e7fc4d9a943caa05314adb88807a8 : FAILURE in 34m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/69d3b87968fb47f9a7a1a119392f3f71 : SUCCESS in 34m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0b3eb68b4ba44583a03203e20c1e3556 : FAILURE in 40m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/691aaa47841a402a9396495859fdcf37 : SUCCESS in 54m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2408a668a1dd4de9a14179783faf52ea : SUCCESS in 54m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/24e8efda69564bffbf673b1b04de7295 : TIMED_OUT in 2h 02m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"325276fa41d488b2ee36ac9a05610c8668914de6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 18:18:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2532c4d90c8df18eb888137c99aa49978590a3b4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 19:45:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34e809a1cb92e4bb69d6cea04312379ff7514f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:53:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a25bba6cf4742e489f374a7603057a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa5827478d64072995291a77e3abd7f : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b5dc316077c4eca97cb6b7a71b7a95b : SUCCESS in 13m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe0d12aafa1a4ec5acad39874ec19c4d : SUCCESS in 10m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf7aabb652084340b5a0e2cc4ada32e8 : FAILURE in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14fd540d1534c6bab9834930e745d5b : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc1afb25844741d4a07ba2bacc78d110 : SUCCESS in 12m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0995373dee1947689db20806dfe4fcc2 : SUCCESS in 2h 04m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0c339c3fa594a149cab66f006caab7c : SUCCESS in 2h 30m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d83df59097634c9a8827df785debcbf1 : SUCCESS in 15m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9c1d023ac62e4049b535e7d47010fe69 : SUCCESS in 15m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/273f396fcca94c7bb9fd06b8922b89fe : SUCCESS in 1h 59m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f6dc74a92d7e455480eb4dedfb59503c : SUCCESS in 58m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bb1ae95b8de3458095d55ebc8571e6ca : FAILURE in 54m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/142a0cf4276c4066b48f79276e3de98e : TIMED_OUT in 2h 02m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"232240ccbe4df6ba8bfaa5a8a338a0f2dc9b8bf7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 21:05:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"273a51ec13ea9c92489a553efec669a426b849eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 23:08:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b82889f7aee4517ab0fbeda6baa52a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f569de9eb7e549ecb1a8ae38a37cb3fd : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f03f146251a44b3881eb9b9955d358c3 : SUCCESS in 10m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d8dec663b32429db973c85fc8ee3b52 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99bf4555854244a1b0365eb57b49d58c : FAILURE in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a1396cf6a941b1b1aaa406c2f5cdbf : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5075a30320384311a95bb0c357d4903d : SUCCESS in 12m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/add248b64ca843bfb5a50c7b2c8beff7 : SUCCESS in 28m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3546787dc17849edb902001f879f76b5 : SUCCESS in 23m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cda8a8f0eabd4725acdd7fcc894d95ea : SUCCESS in 26m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ee383197bfe54550a3625860061d6c1f : SUCCESS in 15m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6e5b53ed68c24c8098b18b47e36cfca5 : SUCCESS in 15m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/37144791368c4b6d9993a0651e709b15 : FAILURE in 31m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/91f997e9f68940609c2b941bd1ce2d44 : SUCCESS in 37m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/883686b2c27a44499b031e710281adf0 : TIMED_OUT in 2h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1c38652040fdda1360c8c1a4acfd7787c82e9c6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 06:06:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffa218967fc1540f91d22a50f87ecc5a1da02b45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 06:08:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dd12980bfbf5c3f038bb278398aa8d097468629d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 08:15:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/590a411b48b648cc8bfdbaee799a3b01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86260f83de34447eba78d535657ac21a : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb3573eb021439fba67300b42755bd1 : SUCCESS in 14m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/254f10b19adc455ca0b733538c70cc24 : SUCCESS in 11m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/862ab3b7f17248dab0fee9899669f5ca : FAILURE in 13m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bfdb9fa2d146b49c6919c416b57bfd : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98a0ca7356d54ea9b8e8d01edb3c04dd : SUCCESS in 12m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0ab4558121114677abdeaef30cd020d4 : SUCCESS in 31m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5cdbb085d9a4b559e186b27301618cd : SUCCESS in 32m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fecf4d4170a41ea86703c4286ae5b1c : SUCCESS in 31m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a5cd23afdbb845e0b2cea7389fb1849d : SUCCESS in 31m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bfe3d09cc0f14d2b8f2dc7f6670d04aa : TIMED_OUT in 2h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e42e9b9c2a92e82dd8a1c99e740e4f5e97526909","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 16:08:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"81d5852a7af3e3aadbfae6a9d92c94d349309aff","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-05-02 16:16:43.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9b9a6c04710b98dc34f3b26d36b75f42cb80a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 14:15:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"854ecc6e55a366199198c48a5fd43dd1bd5c6e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 15:03:25.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/948390 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"01bcf634375934717b03e7aac28edbba4f6d7b70","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-05-07 06:08:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09c9ae4f93e82c4a5a74ad0e011de1066060b89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 07:31:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e8e787bc9e4406b55f1cc9d8b462b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6784e5614b6487793caf87a263dd50b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ce3de0e6a3c40fe9243883e47ae48f1 : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/965525e59eeb40e0aa3edb7f3f0c8ebb : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccd3002ce0404fd4898db6cb826ddd22 : FAILURE in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b718e68511c84ae491b7088ade9544af : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b759a1ae18040749c2eb3f8c77ad131 : SUCCESS in 12m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/14edbf6d862d42f2bfe637d3d6913c04 : SUCCESS in 18m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/977ee3a4ea094eb6a672576275e3af0e : SUCCESS in 14m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71c2aaa4abd24d1691684e8ebc83e9fc : SUCCESS in 32m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7d61d1b2aab846bb9893a9db73a8ffa1 : SUCCESS in 24m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a0e005f63a143808e0edc3896521ad1 : FAILURE in 1h 21m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f60d718c1c35cbfef5fa47656d95ae20862c9488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 07:31:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9cfaf705103d03fe7cad96831b1545b6b544fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 08:13:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ac0077abca747b2b48afc17bdd73fc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fff147f6e84233aef31cda577eb60a : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/458a09acb31d4a6899cdc19e6fe17c71 : SUCCESS in 11m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7264f429546845d3bf17e7f5b8475e61 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fcd75b58224a7c837e90018af3c546 : SUCCESS in 6m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7da76110bc19467c8010dff0a9377759 : SUCCESS in 36m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d2c2d48ea084cdeab6d6a500bd2b533 : SUCCESS in 15m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fe4b5418626447bb6245202d4359f0a : SUCCESS in 32m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b163d3bc3a754bcf82d0667e6cf9fa1a : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":2},{"id":"5a4e9f55f76b51ff5212ace29d63277515defc59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 08:13:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0e5bc5c9201b7a08f104e46eb54882c122dd84da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 08:14:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b4b4ae016ac4cfe910bbf57dc8baf37\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af3401cdf094961a49154ff6d080693 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15aea22d35c12070f649a80a0d767bc87d793065","revisions":{"1457a5b264b54c6ddf6604b406ce511be23b70c0":{"kind":"REWORK","_number":1,"created":"2025-04-28 06:04:36.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/08/948308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/08/948308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/948308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/948308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/948308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/08/948308/1"}}},"commit":{"parents":[{"commit":"9a50e05ebfceb0129aea4ca684e0ddb709393c8a","subject":"Add allowlist and denylist for action providers and actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a50e05ebfceb0129aea4ca684e0ddb709393c8a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 07:36:34.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-28 06:00:56.000000000","tz":120},"subject":"Add legacy action group to list_opts","message":"Add legacy action group to list_opts\n\nThis is needed for documentation to be\ngenerated for the config options in this\ngroup.\n\nChange-Id: Ic536029e30ebb2f620e4c3f99a53a7d400fbb9cd\n(cherry picked from commit 9fba42f3e0a9348e61430ab6769e2356df30e0d0)\n(cherry picked from commit 953e5b9a28cef54421a24fffa4e1e83423578a0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1457a5b264b54c6ddf6604b406ce511be23b70c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1457a5b264b54c6ddf6604b406ce511be23b70c0"}]},"branch":"refs/heads/stable/2024.2"},"15aea22d35c12070f649a80a0d767bc87d793065":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-29 06:08:11.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/08/948308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/08/948308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/948308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/948308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/08/948308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/08/948308/2"}}},"commit":{"parents":[{"commit":"573aefce28addaa12b4527915cbb43be852f23db","subject":"Disable some tests that are either failing or irrelevant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/573aefce28addaa12b4527915cbb43be852f23db"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 07:36:34.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-29 06:07:54.000000000","tz":120},"subject":"Add legacy action group to list_opts","message":"Add legacy action group to list_opts\n\nThis is needed for documentation to be\ngenerated for the config options in this\ngroup.\n\nChange-Id: Ic536029e30ebb2f620e4c3f99a53a7d400fbb9cd\n(cherry picked from commit 9fba42f3e0a9348e61430ab6769e2356df30e0d0)\n(cherry picked from commit 953e5b9a28cef54421a24fffa4e1e83423578a0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/15aea22d35c12070f649a80a0d767bc87d793065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/15aea22d35c12070f649a80a0d767bc87d793065"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
