)]}'
{"id":"openstack%2Fmistral~948355","triplet_id":"openstack%2Fmistral~stable%2F2023.2~I49884c404d7187c612de434f1d5f1698588adebb","project":"openstack/mistral","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-29 08:15:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I49884c404d7187c612de434f1d5f1698588adebb","subject":"Disable failing CI tests","status":"MERGED","created":"2025-04-28 20:01:45.000000000","updated":"2025-04-29 09:03:55.000000000","submitted":"2025-04-29 09:02:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948355","meta_rev_id":"cbc3985ad1e49c96af8a30ffe0d6940642a82150","_number":948355,"virtual_id_number":948355,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 09:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-29 08:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-29 08:15:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-28 20:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d13c623e41fdc2203e072b7ad38ec883e23b918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 20:01:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5b10af1a30481aa96ab5d569883c5f646e63b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:03:22.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/mistral on branch stable/2023.2.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/mistral/.zuul.yaml@stable/2023.2\", line 108, column 19","accounts_in_message":[],"_revision_number":1},{"id":"291fadb98006263cd26ac3007642941801d86b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 06:05:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cda152af732e6b4a5cdaa3893a5d99ef1341d20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 06:50:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa2ad1106b88436995c431e0477763db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a8fe1522b74477bd92d20dda3d81c1 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/804c9d7181e34dcc969d956744aca136 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4689fe9910084d0fa9e87b82789f9c90 : SUCCESS in 9m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad5db089c4d942e2b114a8f1752930eb : SUCCESS in 9m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85bd6014b7cb4eb3a898fbd9576480a8 : RETRY_LIMIT in 1m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a231a304ef444ac5b83a98e35f26d1c0 : SUCCESS in 6m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cd6d497ab12d4992b8e967eeaee3a4f8 : FAILURE in 24m 35s (non-voting)\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da36c42cac9a42799d8d1b509a7e76d3 : FAILURE in 29m 39s (non-voting)\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f1fd5e397194db3934c2345f0c957de : FAILURE in 29m 32s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/645162be42cf4341a5980cfbb0c91a31 : FAILURE in 27m 26s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e947a199bb47479b91ff2d46efe98f34 : FAILURE in 24m 09s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/71bbddd4814f422ead79b61a2528f26a : SUCCESS in 40m 46s (non-voting)\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df3da7bf39364f8390ca4eb8787d4838 : SUCCESS in 39m 52s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1943a952621b41f3b4c74da1c9524331 : FAILURE in 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f73ab8f012166aa8fae4944e7584df0b19741d8f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 08:15:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"07587241402dbb770be295b164e82f38683c07fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 08:16:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99eb307c63e366b288d6422df2d65d747584de43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 09:02:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6323030beae34e30956e30121468f57a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce46cd00416435f93e21ff50dd73e41 : SUCCESS in 2m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b2847d59971466bb0b7213ffe65e058 : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efc1cb6a30cb4c02b6b64e8274a45819 : SUCCESS in 8m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8cfeba8e7b64cf4b4fe29708059defa : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3980eec6b4d474db03655578c834097 : SUCCESS in 5m 34s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7214a2987e7a423190359baefc276945 : FAILURE in 21m 07s (non-voting)\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a88bfa1869042c3b2d096491495ef38 : FAILURE in 26m 05s (non-voting)\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9a65477557d4924b3c36eeecf5f39bf : FAILURE in 19m 01s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8da3c3788f3a40b5ba4f980c3e385bff : FAILURE in 20m 19s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4262006c7e1049df876081f337097921 : FAILURE in 28m 04s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/09adb290d1394c5588694b35d4e9f217 : SUCCESS in 36m 22s (non-voting)\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2539de9c2f5848fd990713478c640135 : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10c5116b275f23fad360ed3efc12202fc9dbfdc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 09:02:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cbc3985ad1e49c96af8a30ffe0d6940642a82150","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 09:03:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31791ff1fbb34baf8afa67f765fe13b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adcf5323a4a4e9cacf56a8c8de83fb0 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ae646cca7acd4933789c32e4741fd5941feb1d0","revisions":{"d2a92171a7ca37540835de1bcd6de018f949df55":{"kind":"REWORK","_number":1,"created":"2025-04-28 20:01:45.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/55/948355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/948355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/948355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/948355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/948355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/948355/1"}}},"commit":{"parents":[{"commit":"7baff017d0cf01d19c44055ad201ca59131b9f94","subject":"Merge \"Fix test not waiting for action to be created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7baff017d0cf01d19c44055ad201ca59131b9f94"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-28 19:58:39.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-28 20:01:23.000000000","tz":120},"subject":"Disable failing CI tests","message":"Disable failing CI tests\n\nWe think it\u0027s important to have a clean tox.ini for the 2023.2 branch.\nSo let\u0027s disable the CI for that purpose.\nWe will try to revert that commit right after.\n\nChange-Id: I49884c404d7187c612de434f1d5f1698588adebb\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d2a92171a7ca37540835de1bcd6de018f949df55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d2a92171a7ca37540835de1bcd6de018f949df55"}]},"branch":"refs/heads/stable/2023.2"},"1ae646cca7acd4933789c32e4741fd5941feb1d0":{"kind":"REWORK","_number":2,"created":"2025-04-29 06:05:01.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/55/948355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/948355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/948355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/948355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/948355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/948355/2"}}},"commit":{"parents":[{"commit":"7baff017d0cf01d19c44055ad201ca59131b9f94","subject":"Merge \"Fix test not waiting for action to be created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7baff017d0cf01d19c44055ad201ca59131b9f94"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-28 19:58:39.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-29 06:02:51.000000000","tz":120},"subject":"Disable failing CI tests","message":"Disable failing CI tests\n\nWe think it\u0027s important to have a clean tox.ini for the 2023.2 branch.\nSo let\u0027s disable the CI for that purpose.\nWe will try to revert that commit right after.\n\nChange-Id: I49884c404d7187c612de434f1d5f1698588adebb\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ae646cca7acd4933789c32e4741fd5941feb1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ae646cca7acd4933789c32e4741fd5941feb1d0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
