)]}'
{"id":"openstack%2Fmistral~948356","triplet_id":"openstack%2Fmistral~stable%2F2023.2~Ib7f705768833dab71b32e474d96474ad50a721de","project":"openstack/mistral","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-29 08:17:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ib7f705768833dab71b32e474d96474ad50a721de","subject":"Revert \"Disable failing CI tests\"","status":"ABANDONED","created":"2025-04-28 20:01:45.000000000","updated":"2025-05-02 10:08:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ade7248cc7cbb13263601250b379211176eea258","_number":948356,"virtual_id_number":948356,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-29 06:53:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-28 20:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-02 10:08:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"3f587baa5b250e75822ef482695bfa0857fb15a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-28 20:01:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"234fedcda83a85953a8c5fc4d25b2ae8e77f6c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:03:29.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"5fce63975e33c61530222d204ce10f6223421f9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 06:05:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5953109105c239d3d440bf0481309bd57658f35f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 06:53:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f9045f079514c929e835a0a138ecf4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8accc22abd86485fb0b6dc86a678003e : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c742250fde1a45aa9ca28f204c98b102 : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e0c46495ff24c25913a689420b08dc9 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d21870c849d4a2c900fc6fe4706104d : SUCCESS in 10m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/262fb0da89e94d17aa0aff64ab077a1e : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd68f8a98e1a471e884c6dc7521dc832 : SUCCESS in 6m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/041696bad2b54441b00e22556d9376ee : FAILURE in 33m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9282a69072b44cfe9169e7337aa10e30 : FAILURE in 28m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7c537674af14f249b41830be87855c3 : FAILURE in 24m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5994af9d8aaf4636bfe72f726828a53c : FAILURE in 24m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/64e9efd4ada943b2a76cf01b1b2a3b48 : FAILURE in 24m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3d90a69c3654406196d488a1df99073a : SUCCESS in 42m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d4300ee2b65d40f0a55edadcf4fc8d4f : SUCCESS in 40m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/43555f9456444bf4aa29b25a851cfa55 : FAILURE in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f457396bb53933661570dd1748ec68c5d10add3c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 08:17:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ade7248cc7cbb13263601250b379211176eea258","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-02 10:08:45.000000000","message":"Abandoned\n\nstable/2023.2 branch of openstack/mistral transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16395a26e2f15799bf89953c4748427eba8a0da6","revisions":{"4444433b52413e426c7f0faa317850499da3ef8d":{"kind":"REWORK","_number":1,"created":"2025-04-28 20:01:45.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/948356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/948356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/948356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/948356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/948356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/948356/1"}}},"commit":{"parents":[{"commit":"0b2e7b50aba1c1141c286b531df3068491e96e65","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b2e7b50aba1c1141c286b531df3068491e96e65"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-28 20:00:07.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-28 20:01:31.000000000","tz":120},"subject":"Revert \"Disable failing CI tests\"","message":"Revert \"Disable failing CI tests\"\n\nThis reverts commit be1665311b993321c6e34dafd8fc49f3706c54e8.\n\nChange-Id: Ib7f705768833dab71b32e474d96474ad50a721de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4444433b52413e426c7f0faa317850499da3ef8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4444433b52413e426c7f0faa317850499da3ef8d"}]},"branch":"refs/heads/stable/2023.2"},"16395a26e2f15799bf89953c4748427eba8a0da6":{"kind":"REWORK","_number":2,"created":"2025-04-29 06:05:01.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/56/948356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/948356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/948356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/948356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/948356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/948356/2"}}},"commit":{"parents":[{"commit":"9cbf95fe9bfe739cdddbbc8184de3449545ea03c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9cbf95fe9bfe739cdddbbc8184de3449545ea03c"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-28 20:00:07.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-04-29 06:03:50.000000000","tz":120},"subject":"Revert \"Disable failing CI tests\"","message":"Revert \"Disable failing CI tests\"\n\nThis reverts commit be1665311b993321c6e34dafd8fc49f3706c54e8.\n\nChange-Id: Ib7f705768833dab71b32e474d96474ad50a721de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/16395a26e2f15799bf89953c4748427eba8a0da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/16395a26e2f15799bf89953c4748427eba8a0da6"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
