)]}'
{"id":"openstack%2Fmistral~949002","triplet_id":"openstack%2Fmistral~stable%2F2024.1~I57cd468cba227cc8cc7960e84e9a111217e80cad","project":"openstack/mistral","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-24 14:20:43.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-05-22 13:52:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I57cd468cba227cc8cc7960e84e9a111217e80cad","subject":"Add allowlist and denylist for action providers and actions","status":"MERGED","created":"2025-05-07 09:14:40.000000000","updated":"2025-05-24 14:21:34.000000000","submitted":"2025-05-24 14:20:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949002","meta_rev_id":"ebdeda46eb3798b560e5f810e4451e0696c8a2fd","_number":949002,"virtual_id_number":949002,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-24 14:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-05-09 07:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-08 09:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-05-22 13:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-05-22 13:52:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 11:17:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 09:47:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-08 09:47:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-05-09 07:42:20.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"0b463e5491d9eb2539f2cb123bc1b6301fed6f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-07 09:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c456ba3d2f5b0b47f1c8e2005c97d94fc7a2c187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:17:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd55792374ef464d86cf3a4ce37d0b40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03c734cc9574ea4a710e35548cea2c3 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73cadf97b06845cd88e12268090204f5 : SUCCESS in 10m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2b7c6968e484d6593fe081c2ac50b5c : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66de2a91546d4600a59b6b1f0ea7b279 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47dda28f51194decac9b6c466012d4d1 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6354cfe7a79645c8bc05e2af3b1c83e1 : SUCCESS in 16m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/23387e9ddeef4f778e30617946f7c718 : SUCCESS in 29m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcaeb8a57dec4b90af99f788574cb931 : SUCCESS in 38m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd2d2b036a614d7181ba6cdc3fecdc91 : SUCCESS in 27m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/be9d5580be744ca4ba625644b398aae1 : SUCCESS in 36m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/87894821301e408d836288c7ea7cabd7 : SUCCESS in 39m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a54af67117294339b4deee43e1b0ad35 : SUCCESS in 40m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d3abaacd29f4762a0575a2e2e5ffe0a : SUCCESS in 45m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c374b28cb7d64315a21ef987fcf074a8 : TIMED_OUT in 2h 02m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c7a5c60ab25895743d68c8b85dab3ac52de0013","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-08 09:47:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"819afd508069d97af303cbb40f053d4d3ce38af0","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-05-09 07:42:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6301f926f281662edf80fd1ccc388bddd396ed4b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-05-22 13:52:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6908e36cda67d0aae6ed4be1f64fcd074193623f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:53:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3850a3bc2ca75d5d7a4812e989b664f08ef8245","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:45:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/565f8692f6264e088f6cb0114855dd70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c225068d6f14009aeef1815976ea1b1 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66454f5201854e808bed7113ddd2f901 : SUCCESS in 13m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f385b58be134c2785d9d8b9eb90fe91 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c21e25b68f042588f875098b6bfa9ac : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34d14b904ac146aa9080dff9700e7c73 : SUCCESS in 2m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f5875711325a43c291b2e2743c685981 : SUCCESS in 47m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b8d9162663d444a92d92b88872186ce : SUCCESS in 32m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d4ce2b335304b69a098c38bfdc892c4 : SUCCESS in 34m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f4517e7def3c4713b21901c0398c60cf : SUCCESS in 37m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e87160b979bd477f99bb37274b729318 : SUCCESS in 38m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0fcb4d9c2f984805b1c4ce066ae3bd62 : SUCCESS in 44m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38f050ce533848af954dd9f7b877b808 : FAILURE in 40m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4dcabc72030c01217b438427b8a219b39dcbf8f6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 18:11:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10d217783b1fd33b76a401d4d23c476aa18cfabc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef464ccf3ce94ce79d8277f4607d78e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5654aff33d437985f00589218beaef : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cf00f5e02a54c42890385018fa06ead : SUCCESS in 14m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4b8deef3115419d8069e55ec24f71b2 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d6c162e4c4443aa15477fb49268f4f : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50bebf25389f4152a3aefb3948983d23 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60ffeefb8ba94e1fbee33e1e58401576 : SUCCESS in 13m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2c3160701ee14723a07b8dae4fb61fbd : SUCCESS in 33m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd4bfce12d1547dc8077adb615344a6c : SUCCESS in 38m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12700e079b60461ba21e562d0b7636ed : SUCCESS in 38m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c62f5b24379f4d05bbbb7bbfcd1bbfa7 : SUCCESS in 39m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f8695337107446bdaf234f08d195d55b : SUCCESS in 34m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/745d44082b0d490f97a89dad394bc658 : SUCCESS in 45m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/65150d7f3e7f4f1385df966da1b4686a : SUCCESS in 47m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9833cf5153024b4f9a2da6bf50aa4a66 : TIMED_OUT in 2h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"659c465f985a62cd6cc4930365ef4f83cb842cc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:14:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c55229f46f463c3b54e20395fa973305fd5c4263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 21:16:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92d99b9b090a40f08dceecf78641ca47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e079a3f92ae4b1f9a7ea4dab2593d3f : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c59c63ce23f643b0b8422079d25c5f6b : SUCCESS in 10m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e458973f7f67427ab7a1f1b5a1541c55 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e82bc22676442ea25df8a49c78ac57 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/736ddf4792db4db691a86f588920b6e3 : SUCCESS in 2m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/43374076df0b47daa0dcca78929ecb1f : SUCCESS in 16m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ddffc73a4bf4402a79af63b9dfdd181 : SUCCESS in 30m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4776af1fab894df789e2e28ff93cb662 : SUCCESS in 25m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/632c1b4bbb6749dd87864e05fcd3a581 : SUCCESS in 25m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9a3f5ef34f3443d7b372612cc726ff9a : SUCCESS in 42m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5d18898fe9184b56846a803c9fe80840 : FAILURE in 56m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ba9fff53a638433dba57341f9ce4fce4 : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1441150a5f10876002cd8b920ea27f8b522d0981","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 21:34:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7be52a2626845959031b97b2eeff6f37cff8480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 23:43:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f78681d253a9414b861322a11fb32dbf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86bf4d8a52ab4f77aac56624a3790a06 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/242a97168ddd47b6a81dae738df906c0 : SUCCESS in 9m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cab5b2b184a5476e9ffe2d46543ca67e : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ff1155b01049d481b40eee39f1acfe : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba9e1dfc82894b958815170301556343 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d1cc0ccf5184a7fa4fd582ce4a09d45 : SUCCESS in 11m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ede5d85c513e4453ad93d5d062f5f6e9 : SUCCESS in 26m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06443a95c2e64dffb52644f5b1f08b54 : SUCCESS in 37m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b658c58485544f1b78c0630adae957a : SUCCESS in 41m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d2e8a274ab51467689ff4ab49cc221ee : SUCCESS in 39m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/758f00b4314d432e9884203c8eb222e9 : SUCCESS in 39m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/38b1b6faaa9f4e7d8c5e81c2534af86e : SUCCESS in 56m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d85d7c675dde422e9a09e134b9878cdd : FAILURE in 55m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/37f240b370e64f489c057ac1c826bbff : TIMED_OUT in 2h 02m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7df4d08fba669e6195c13991f77fe1903083b4c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-24 08:05:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"427061f6368d1eea7a10faa94b66c73f64cd1b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 10:15:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467028f5bafa4560bce6c761b21cb60f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df742e947d1c455b8d01165d2008a306 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3606fb3695004d30bd38365144bca456 : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d08385aac5a74d75ba9eea879368ea92 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4b9eed9b8840f693617691c651e7f9 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f365d51c8e134c6da436bd86685c91dc : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a378aa1c9bfa43e195fb4eb31a038494 : SUCCESS in 16m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/583fbf0e6cc040468984b487a796b15a : SUCCESS in 40m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f05968fe9df64b5faf57fac06acb6ac8 : SUCCESS in 41m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97ec698eac1a417fa3bfe99348c3ce9e : SUCCESS in 38m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2425e0af6b3e4fecbfce0fbfeeba08f4 : SUCCESS in 30m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/56c2b6f87d95463694fc27695d82d6d5 : SUCCESS in 29m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/444edbba73134686a06da907f1cc8b0b : SUCCESS in 1h 00m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fe187c7df9e54c6b8b0976afbb5ac2c8 : SUCCESS in 28m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0a5f8863517542879ab89ecfec182e0b : TIMED_OUT in 2h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00aa78608fae61a33bc9fffde638d80079400e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 10:15:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21bdc888cce0b8ff235242307c92fd5830f8a154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 11:15:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c64f03effd2e46bfb02c0a0b2a5b308c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356770f12ae84f079fec07296f753cbf : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb493033cf7e44b7a2fa8dc4ba6a6d2d : SUCCESS in 10m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c05322bc383b448eb0bcad42faa332a4 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65be7a6df2d41328f343062417e9162 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45e0a128ea7c4593996712df2a3a1c2d : SUCCESS in 1m 43s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b5acf95a8f50420d83604f8d1703c380 : SUCCESS in 25m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bb60c570059424583704574cb0d47e7 : SUCCESS in 27m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a1065cf0b2d4e63b0ccc82c88ad1ffa : SUCCESS in 28m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ee83c04c8bed427e84cf4018cafe18ea : SUCCESS in 25m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bce0f7c89aa64ea9ba0803f51cd2008a : SUCCESS in 37m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/37a6b6769e3c491d96cd12a07b0fbd22 : FAILURE in 40m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7b96b71727834707a05a60c24f3c0123 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5fdc42457bb709586b68fe83c71e55de2b97bde4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-24 11:15:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6806d9902bbf69470fcd656d321e176857e3ba15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 13:22:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed10ce50e78349a1a5b707edc1122c91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020a4b3400d44b66a5f15c855e318801 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb721a21a5f24e92a9b7b53d0da32450 : SUCCESS in 10m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2becd6bd9044375ac1b067dc474b055 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ed2b3f3a4e4c389d4add4499338804 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64e1792daae34afeb1b0132ed41330d0 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce0fe8400bc4071880ba4980d4d763b : SUCCESS in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cf132fa9a97941a0b2c5aab043d734a8 : SUCCESS in 26m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9205e99f1224ef0be5a035485348b7d : SUCCESS in 37m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3067562c4d7b4136b0180595dd61568e : SUCCESS in 38m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4ab3cf69345a41719f3f260f51174d45 : SUCCESS in 37m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/14aaaa797ddc4e6bb62ef4b2be608891 : SUCCESS in 40m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/416bc578dc9d49d89389a05cdcfd62a3 : SUCCESS in 54m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5dc551149ed34b7a9f1c89e6a580325c : SUCCESS in 53m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/270286ff9bf24de3a0c863762eddce1e : TIMED_OUT in 2h 02m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22a433a24069010b5da74a5997fe4e39b8a7ba19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 13:22:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41739c6cd2e8039aeb1d3bb8ddbe704e8f586709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 14:20:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d61fd2b39d4d95a93b8e3759dfedc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ff7601dc3641a4b5b7e29fc85de256 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7c85919d6ee46adbff11ecda6bb9c98 : SUCCESS in 9m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f5b5bb6916405ebefc4452978dc970 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8455a580e05740a7967e98badcc3d125 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e51d14bca0949bbbe2555340dde1f3f : SUCCESS in 2m 33s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c94b4ef817654d3f8b3651f4788c0dd5 : SUCCESS in 27m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd6ada16d98148ff94eb778dac0b9495 : SUCCESS in 27m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e496b81aeb845ba85b9c8703db61822 : SUCCESS in 26m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/39c3b7d4b4b74c2c82d39589607fa219 : SUCCESS in 16m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/93a2c51ab077479e8376b6d0c14d4969 : SUCCESS in 42m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7b1e81f7054a4502a2c2bd0d5341268e : SUCCESS in 31m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0c0591e289a64936b304667a618e1fe6 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"708e722a8a76eba1837c79c457db3e853b629179","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 14:20:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebdeda46eb3798b560e5f810e4451e0696c8a2fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 14:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1348a4064b449f595bdddd760fe550f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6abea7a8754327923e81ec7254c53c : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef43f2ac08004893b7dd45bd9961c55b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"985ed5db198cb3a7b69c33798e3f125e3d27468c","revisions":{"985ed5db198cb3a7b69c33798e3f125e3d27468c":{"kind":"REWORK","_number":1,"created":"2025-05-07 09:14:40.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/02/949002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/949002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/949002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/949002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/949002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/949002/1"}}},"commit":{"parents":[{"commit":"3c16d9d56431f9061db39f9451de1d772a382349","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c16d9d56431f9061db39f9451de1d772a382349"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-11 09:54:21.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-07 09:14:17.000000000","tz":120},"subject":"Add allowlist and denylist for action providers and actions","message":"Add allowlist and denylist for action providers and actions\n\nThis adds allowlist and denylist ListOpt config options for\nfiltering what action providers is loaded\n\nAlso adds allowlist and denylist ListOpt config options for\nfiltering what actions is loaded by the legacy action provider.\n\nThis is used in a security context where we need to restrict\na environment to only load actions using the legacy action\nprovider and then filter what actions is explicitly allowed\nor not allowed to be used.\n\nChange-Id: I57cd468cba227cc8cc7960e84e9a111217e80cad\n(cherry picked from commit 11785d1e31611ebf0b1f11fcb10fcd4dea3efe9e)\n(cherry picked from commit 9e5d9c6ba1cb4a22363b04a548c078600753ae42)\n(cherry picked from commit 9a50e05ebfceb0129aea4ca684e0ddb709393c8a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/985ed5db198cb3a7b69c33798e3f125e3d27468c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/985ed5db198cb3a7b69c33798e3f125e3d27468c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
