)]}'
{"id":"openstack%2Fmistral~949040","triplet_id":"openstack%2Fmistral~stable%2F2024.1~I2b874b7ab2d213a8859b7b4bbd9e546278595c0c","project":"openstack/mistral","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-26 01:00:42.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-05-22 13:53:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I2b874b7ab2d213a8859b7b4bbd9e546278595c0c","subject":"Support fnmatch glob for allow/deny lists","status":"MERGED","created":"2025-05-07 14:38:44.000000000","updated":"2025-05-26 01:01:26.000000000","submitted":"2025-05-26 01:00:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949040","meta_rev_id":"d66dd6ddc42f53dd403794cce9a72ce960ca7b3c","_number":949040,"virtual_id_number":949040,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-26 01:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2025-05-09 07:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-08 11:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-05-22 13:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-05-22 13:53:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 16:25:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 11:28:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-08 11:28:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-05-09 07:41:39.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"dc372fe3f2fc97832d8fa6419aac43b89b4abbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-07 14:38:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2bb69fcb2ccd19e719f9f98493dcefb4dff363e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 16:25:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0ed58212c3e4d02b884f5922eae4cdc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d78a4b1e204407a8f3ce81a373513ac : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83bad3de9c784fcab99145492c85238e : FAILURE in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dcdd467106748f396e2bbf165895f0e : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e877269850470b9454998a7c4ea989 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435b8a84ecd141daa0227e4df66613a0 : SUCCESS in 13m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/384ae8220e144dfdb5f9ce64388e1662 : SUCCESS in 21m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a594c5f540e647cdb8d57fd0dddbeac5 : SUCCESS in 38m 22s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c89b4e851b1c4ac593b5cc2049059ea3 : SUCCESS in 39m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d47bf15d77584be2b08a8cde9008c09b : SUCCESS in 27m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5d1c91983be349869a5dcefb172f36e8 : SUCCESS in 41m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/285605e0542047b58ed8ed493a5ea49e : SUCCESS in 35m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/85a2a9b37f4d491293ba56d6ebea501a : SUCCESS in 32m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d9f958ac8baf447f8991fc18d490dc09 : FAILURE in 1h 45m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7aab8a5e5e210cb2030efeb6d571de543c9b99fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-07 19:12:08.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e5a5c5fa2078e33f8d7caa474df883d1fa23065e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 21:19:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cfa4cd4f9b340b0b131d5f5c32bde3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3144482264c844f9b2f7a33014f25fce : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a99a2ed90e6470f8991e43b3c53317e : SUCCESS in 16m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa545a8b685b45e98f2e236799fc9bba : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0001505f337423491e9903faceb938f : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06380c22feb84b0fbb059c113b5f6d5a : SUCCESS in 15m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5c2a3bfa427b4fc284a39fdf95c570bb : SUCCESS in 41m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c15b3d473ea46e9b63d4266ccbca321 : FAILURE in 27m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/910ac209d6c347168f48acfde344c7bb : SUCCESS in 36m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f62d09163f1d4869b2f54db6b69cd904 : SUCCESS in 17m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bbb82b863d844ef78d355095f5f59743 : SUCCESS in 36m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5dcf937eb1b74872be101ef303d80309 : FAILURE in 54m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/102991ccbec740ca8c340d6f33b80740 : SUCCESS in 53m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/35285de69ac444e29edf6d708df7b90f : TIMED_OUT in 2h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2bc89c14173f7805b202adbd38c85bc9875612c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-08 08:20:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd25125378092606c0d0bd3398d2b171398938c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 10:27:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4993aaf159945c981131da44c15c36e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bca0792444b42299cd3fd779034efd2 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a864bb6551e2482083e6541d677cde14 : SUCCESS in 15m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/969bdb182f794146870f2ac7542eef62 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb0c40582ab4c6bba2ece24dfe3fb3b : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e880f6843146adb8dd09da31967f42 : SUCCESS in 12m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bc56fd1a61614a36b9a621f59405dddc : SUCCESS in 38m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7cf9328dffe407ba1ee3950ac3d78dd : SUCCESS in 31m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/434d1ea80ea1496eaf790f3a27f5ae47 : SUCCESS in 31m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f7c48a0f8dc5452ea37922991947188a : SUCCESS in 17m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4f9f7f1aa52743fb9f4211f64899b8bd : SUCCESS in 30m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/67e5d216c31740daa5fcdc328665501a : SUCCESS in 55m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4ec48eb178c643fc90072183491df9d3 : SUCCESS in 53m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/473ed70f0f06450b8592158ad29fa881 : TIMED_OUT in 2h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e00fec7e941a630a737889eb5a662469a26a140","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-08 11:28:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dd00a73643f40831992368400d41e7cb28f6cf51","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2025-05-09 07:41:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d65f4601f1b8041effd82c7326c2838a76bc9305","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-05-22 13:53:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef80d06850f3ad0584ebe33ed0570880a187c11b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:53:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"88c4c509bbe5011a544170e7ea2ec7ce70d9cba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:45:49.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/949039 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"c6e47a7f2abdb2f4d03f20016a09d49931ea7bd4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-25 21:44:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3e283f83e071e088875fff9d2eb4d2032d0fbd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-25 23:55:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94914cf022d84f3d9060165c7fabbdc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c70546d7a44bbdac22dbab81536293 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66961ef8efac40faa1e178a62b5a2ce5 : SUCCESS in 14m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7e2dd43fa814771b0cb342a24c879d9 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674b69e9a98045bba0334cac21bac436 : SUCCESS in 9m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/914c1cecd77d4515acc163c9ef7ae62a : SUCCESS in 13m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cb6a43c36e4d477bbbf4f9b4d0e14fd2 : SUCCESS in 37m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87a2147aa9ce442fb4f487028d091616 : SUCCESS in 27m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/465fe0fa41bb461dafa379fde6ef729e : SUCCESS in 35m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9cd84f89fe6748978bce59d87a6b8d46 : SUCCESS in 24m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/973b7f17f8294817bdc912505d300cfb : SUCCESS in 24m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/af52986f741e4368bc7967ab926915ce : SUCCESS in 31m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4ae41135944145d1b9610842726c4e16 : SUCCESS in 53m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7aba4ae27cb54e4f854d85bedb90d781 : TIMED_OUT in 2h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd7807e654c52e169c57f56278350f03dab93176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-25 23:55:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9a157cd6094f44cb23002182d7664c8f0c820b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 01:00:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/262244655d3440f785a1ce7ab3d9a741\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2885282ba8649ad97f3d02f13204eec : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59cb3287dada4c5297a645f18942e388 : SUCCESS in 11m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b084dca67a74ca99010143ee7e4b3f5 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b458a6e77f4f4577b8b08838cb486fc2 : SUCCESS in 5m 56s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c902cd558ca04e27864083289b9407d1 : SUCCESS in 24m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0efb10f5ba1f4df9beea15b84ee4cd03 : SUCCESS in 17m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f0b26d8510249399e5e1b5ef5b5c685 : SUCCESS in 24m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7a082d01bda94edb8ea33a01f1fc28fb : SUCCESS in 36m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b69de29a4a0f4155adf46c69855cf955 : SUCCESS in 24m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0e5a710974a349b78d2d87cc480d4ed6 : SUCCESS in 58m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9986c4be75084784adc7d5fd5e0d0dcb : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bc590ba1c820775a61bc92db5935818809d3fb37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 01:00:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d66dd6ddc42f53dd403794cce9a72ce960ca7b3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 01:01:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2418191cd852406aa27b9abfbc62bfce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9092c872fd694b16940c5d6946ced0f0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82c573e256fde793481036e57cd52289e75bd67b","revisions":{"eceb1f829caf202b2eb4d7892b8476cb5d150c1a":{"kind":"REWORK","_number":1,"created":"2025-05-07 14:38:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/40/949040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/949040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/949040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/949040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/949040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/949040/1"}}},"commit":{"parents":[{"commit":"9a961a83588591a1a1cb0a62cdd9c44532162ebc","subject":"Add legacy action group to list_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a961a83588591a1a1cb0a62cdd9c44532162ebc"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-07 14:38:35.000000000","tz":120},"subject":"Support fnmatch glob for allow/deny lists","message":"Support fnmatch glob for allow/deny lists\n\nThis adds support to the config options introduced\nin [1] to use glob patterns matched using the fnmatch\nmodule.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015\n\nChange-Id: I2b874b7ab2d213a8859b7b4bbd9e546278595c0c\n(cherry picked from commit 371a65fc875e82deef4215d2a98a707b67fa81b7)\n(cherry picked from commit 32e370470dc379b4359d7198e0eeb12568350a08)\n(cherry picked from commit cfc1b44046c242ad3679a60cc2abeb00f02e6658)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eceb1f829caf202b2eb4d7892b8476cb5d150c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eceb1f829caf202b2eb4d7892b8476cb5d150c1a"}]},"branch":"refs/heads/stable/2024.1"},"82c573e256fde793481036e57cd52289e75bd67b":{"kind":"REWORK","_number":2,"created":"2025-05-07 19:12:08.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/40/949040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/949040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/949040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/949040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/949040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/949040/2"}}},"commit":{"parents":[{"commit":"9a961a83588591a1a1cb0a62cdd9c44532162ebc","subject":"Add legacy action group to list_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a961a83588591a1a1cb0a62cdd9c44532162ebc"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-12 08:54:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-07 19:12:08.000000000","tz":0},"subject":"Support fnmatch glob for allow/deny lists","message":"Support fnmatch glob for allow/deny lists\n\nThis adds support to the config options introduced\nin [1] to use glob patterns matched using the fnmatch\nmodule.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015\n\nBackport note: Removed type annotations in this backport\nto support Python \u003c\u003d 3.8 better.\n\nChange-Id: I2b874b7ab2d213a8859b7b4bbd9e546278595c0c\n(cherry picked from commit 371a65fc875e82deef4215d2a98a707b67fa81b7)\n(cherry picked from commit 32e370470dc379b4359d7198e0eeb12568350a08)\n(cherry picked from commit cfc1b44046c242ad3679a60cc2abeb00f02e6658)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82c573e256fde793481036e57cd52289e75bd67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82c573e256fde793481036e57cd52289e75bd67b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
