)]}'
{"id":"openstack%2Fmistral~951361","triplet_id":"openstack%2Fmistral~master~I7c6eb396ec99d7f5957bcaf0548a8e1c817e7e2f","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-31 23:32:39.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-05-31 22:33:18.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I7c6eb396ec99d7f5957bcaf0548a8e1c817e7e2f","subject":"Replace pkg_resources","status":"MERGED","created":"2025-05-30 04:46:32.000000000","updated":"2025-05-31 23:33:24.000000000","submitted":"2025-05-31 23:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951361","meta_rev_id":"2922e8516ae51b8586517ed36bc31101d21d7e97","_number":951361,"virtual_id_number":951361,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-31 23:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-31 16:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-05-31 22:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-05-31 22:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 05:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-31 16:36:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-31 16:36:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"fa67fbb3f70443561fcb22cfe6ad7a05499074a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-30 04:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa9fbe9494196c961f4622465f67b4a482813335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 05:29:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8140ce0529b2495097ceafe863e2934a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e90b76e9943347209b0f2015a3ecb8a7 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a6da422346f465ab4532057756d222e : SUCCESS in 9m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca700778f64a4da0ac5a8ac9a55e6a53 : SUCCESS in 9m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48a6b5c42ddb436da2e43c02b932a023 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c047cdd8b240f59c44ea99d8139552 : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5336e6620a8a4371b59d4d06a24f0527 : SUCCESS in 13m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5b909236e5eb4ae7980b78b06d61a8a6 : SUCCESS in 24m 20s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8b146479d9744558a61baf9086c1cabb : SUCCESS in 35m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04951a972c8b45c49d582b00589d44a4 : SUCCESS in 24m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f82ba4be5f53472eaf6cfcb57012e3ac : SUCCESS in 38m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8337e670a9484e138cfefb395060ec88 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"387347878b2841a5176a8fd0775f7823a7e7b0ab","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-31 16:36:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f475d6d13a5a8f050d285a472f32c336fd02c096","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-05-31 22:33:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16542a2add971603966432b3fb13199db741b6ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 22:33:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8eff35a7deef18ce153c5ca2c77cb458e61baf7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 23:32:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fd0cbc60d9401daea72872ce6edb71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/021eb5c9c140475b9d12658f600a166d : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f24d529ab9a40008e723e8dbe00a8e8 : SUCCESS in 13m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0de3f49d036244e8aa45f24e263d0cb6 : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defa6ad6903246c192c0929d0839a7d9 : SUCCESS in 8m 16s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0e6e1fdb61ba4ec4a1176f6b26860fac : SUCCESS in 36m 59s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/420086ecb9d74b819aed5b3d5799e6ff : SUCCESS in 36m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cef5a8f1511a498588937225f5c43602 : SUCCESS in 24m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be6e9fdbae1e449093556b07d9ad89b4 : SUCCESS in 23m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d229b0624e3b42a49cb271f3736cc3d8 : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3e80e482da24383c0bb04e60a13b4686a74a4fba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 23:32:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2922e8516ae51b8586517ed36bc31101d21d7e97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 23:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14cec8be5ed4434081368f5d5133a8be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be8d5181cf14ec287d950989f3e96b6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21ecea0caf58cfb62e82baa7b7ec80d68e7a32d8","revisions":{"21ecea0caf58cfb62e82baa7b7ec80d68e7a32d8":{"kind":"REWORK","_number":1,"created":"2025-05-30 04:46:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/951361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/61/951361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/951361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/951361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/951361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/61/951361/1"}}},"commit":{"parents":[{"commit":"73ae69e39523ba47df8c790c92c5b589edcb3a57","subject":"Fix tag in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/73ae69e39523ba47df8c790c92c5b589edcb3a57"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-30 04:46:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-30 04:46:13.000000000","tz":540},"subject":"Replace pkg_resources","message":"Replace pkg_resources\n\npkg_resources is deprecated and will be removed[1].\n\n[1] https://setuptools.pypa.io/en/latest/pkg_resources.html\n\nChange-Id: I7c6eb396ec99d7f5957bcaf0548a8e1c817e7e2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21ecea0caf58cfb62e82baa7b7ec80d68e7a32d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21ecea0caf58cfb62e82baa7b7ec80d68e7a32d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
