)]}'
{"id":"openstack%2Fmistral~952624","triplet_id":"openstack%2Fmistral~master~Ie02765348a0d892648fad2b09348c57f1ac0108c","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-03 10:09:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie02765348a0d892648fad2b09348c57f1ac0108c","subject":"Remove eventlet and start using threading oslo.service","status":"ABANDONED","created":"2025-06-13 22:51:15.000000000","updated":"2025-09-03 10:09:04.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8da9498f12a9527f033b87214e81d34a64af3209","_number":952624,"virtual_id_number":952624,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-26 20:41:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-14 00:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-14 21:49:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"2e39082b0009b8932a06e6b95b7a35d96f2267a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-13 22:51:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb755c9d4564a7e84d932f4092a63213dd46f18f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-13 23:08:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff4fcb3102cd1aab1cda92a641831ab5e59e46f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 00:14:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f2a18e137814ac9bed4fb3248958198\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1605ded58184fd7b602a19f48ab0b36 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8860ae767cf84d689d39eb76e1c3c2e3 : TIMED_OUT in 40m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1eee84ce766442f48fc3a9431df9ec0c : TIMED_OUT in 40m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e00bac789be345baac701db56848d59f : TIMED_OUT in 40m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d610ce785b148cca1b19768e5f6c352 : SUCCESS in 8m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/20d66834f66d48548c228470e60abc5d : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a7a0849746a462688399ef6a03c85a1 : TIMED_OUT in 50m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/08f363ee29274a3084bc7f795226b265 : FAILURE in 16m 49s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/118262e4f74e45a28081c90636347955 : FAILURE in 12m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebb2401e86c74ee2837b660a0ece6fbf : FAILURE in 11m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbe2e8a358dc457cbbe2c4c47d850dc9 : FAILURE in 27m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/252e26079a9948b0aaa77ea6e4dd3923 : FAILURE in 26m 27s","accounts_in_message":[],"_revision_number":2},{"id":"520b9fad6be1d6d5bb3e9a40d5b3a81ab7786d65","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-14 21:49:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a84c1437ffb429364a119743bb7ab966977854e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 22:41:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e07de5266dcf4addb67d56f0d111e403\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cdb7e84a8474afebd45647cf0dee30b : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ade978e0086b4de58acace924cf07e0c : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc4430d01dab4517a6b2b1d3741af2d1 : TIMED_OUT in 40m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba70d7cf8fe14d96acd5f369ac862f3e : TIMED_OUT in 40m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57c8790ab8248119780824afc8cfbfd : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cdb52c758f2d4df1abbaed093b41b4e3 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/936bde7e54494a00afe76f4900794c14 : TIMED_OUT in 50m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2e07815aab2a4498ab686a275bfd1a92 : FAILURE in 19m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c6d8bc711c0478bbc1e19a7d0c2e9b1 : FAILURE in 24m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1f26db9df994eefb7b39824b8cfb76f : FAILURE in 23m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8d27b15ae19442a9ac11aa93b7aa4d46 : FAILURE in 24m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bb16ae24207189460708e611418dd1f91424fa3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-26 21:52:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"379bff7c83fb22360004a3fa722fdc4bb7663846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 00:27:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67d833ed2ec84a9ca6d1a79d3f707009\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8571710773fc4b2198c29df1ddf6f84f : FAILURE in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/085de4066dab4f67afce7aea0b3b21e0 : TIMED_OUT in 40m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96c77e3a2d4c465dbd50da75dbe76971 : TIMED_OUT in 40m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1bf98ede6a1e4f12a1a2853c48456ca2 : TIMED_OUT in 40m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163992bc34734da88f5deca3f448dd4d : SUCCESS in 10m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2887f836cf0b46e4899510e4fb77bc65 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35f73e4064124be09a2501ee686fd58b : TIMED_OUT in 50m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c3de5fea39d94f2ab1bd7eb1a97a961b : SUCCESS in 42m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec7a8696144d443ca840026328d808a5 : SUCCESS in 27m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/758dfc444d924fce9fa35f75cc2c060e : TIMED_OUT in 2h 32m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/50bae221fb3a4c63aafac9ecdf7e1020 : TIMED_OUT in 2h 32m 18s\n\nWarning:\n  Comments left for invalid file mistral/tests/unit/test_launcher.py","accounts_in_message":[],"_revision_number":3},{"id":"2c908f5baef0dab80d77ce721c485e32ce0149eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-26 20:06:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0559ce1dbb8d43e9deb6b4709bb3a1e45d228c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 20:41:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a7a3cf5325a49419d700e45e6ad41f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/878ba1ba12084961b3860d9c1e4c01ee : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590e7e27b1804cba86ef6e5c09e70736 : FAILURE in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2a0da009f0c4821a7beb9b44806488c : FAILURE in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d5ce1b389dc4bf88e11f4dc08d1f74f : FAILURE in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcdaf4ee9024bd494b9a91b2fade7d4 : SUCCESS in 7m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/251f2f2ad16a48a3b210191fbd10fcbc : SUCCESS in 2m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2abe6e0df06c4161834d3a49c99ffeeb : FAILURE in 5m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/399b94c1d94c4149aced589368fca16c : FAILURE in 27m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd573e44a86842849930e4e13cf2d13e : FAILURE in 29m 49s","accounts_in_message":[],"_revision_number":4},{"id":"8da9498f12a9527f033b87214e81d34a64af3209","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 10:09:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f4d1a637cdd09032ecad22685d557720a5cfc826","revisions":{"7b2f029de8b9ca931d5439ed04d78da759df8f05":{"kind":"REWORK","_number":1,"created":"2025-06-13 22:51:15.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/952624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/952624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/952624/1"}}},"commit":{"parents":[{"commit":"883d91bdc659ba943d7676843026f4031823ef81","subject":"Allow admin to retrive all tasks and actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/883d91bdc659ba943d7676843026f4031823ef81"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-13 22:49:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-13 22:49:58.000000000","tz":120},"subject":"Remove eventlet and start using threading oslo.service","message":"Remove eventlet and start using threading oslo.service\n\nSince oslo.service merged the threading backend, it\u0027s safe now to move\nmistral forward and completely get rid of eventlet.\n\nChange-Id: Ie02765348a0d892648fad2b09348c57f1ac0108c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b2f029de8b9ca931d5439ed04d78da759df8f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b2f029de8b9ca931d5439ed04d78da759df8f05"}]},"branch":"refs/heads/master"},"221f15d6bece87e73657b43a66dce823ea8d2629":{"kind":"REWORK","_number":2,"created":"2025-06-13 23:08:16.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/952624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/952624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/952624/2"}}},"commit":{"parents":[{"commit":"883d91bdc659ba943d7676843026f4031823ef81","subject":"Allow admin to retrive all tasks and actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/883d91bdc659ba943d7676843026f4031823ef81"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-13 22:49:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-13 23:08:09.000000000","tz":120},"subject":"Remove eventlet and start using threading oslo.service","message":"Remove eventlet and start using threading oslo.service\n\nSince oslo.service merged the threading backend, it\u0027s safe now to move\nmistral forward and completely get rid of eventlet.\n\nChange-Id: Ie02765348a0d892648fad2b09348c57f1ac0108c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/221f15d6bece87e73657b43a66dce823ea8d2629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/221f15d6bece87e73657b43a66dce823ea8d2629"}]},"branch":"refs/heads/master"},"77da9bcbc26bfa49cd96b793aa98a5849d54f247":{"kind":"REWORK","_number":3,"created":"2025-06-26 21:52:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/952624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/952624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/952624/3"}}},"commit":{"parents":[{"commit":"7dec72648d974f67363f2a098698f9d82660d639","subject":"Rework the test launcher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7dec72648d974f67363f2a098698f9d82660d639"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-13 22:49:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-26 21:52:34.000000000","tz":120},"subject":"Remove eventlet and start using threading oslo.service","message":"Remove eventlet and start using threading oslo.service\n\nSince oslo.service merged the threading backend, it\u0027s safe now to move\nmistral forward and completely get rid of eventlet.\n\nChange-Id: Ie02765348a0d892648fad2b09348c57f1ac0108c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77da9bcbc26bfa49cd96b793aa98a5849d54f247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77da9bcbc26bfa49cd96b793aa98a5849d54f247"}]},"branch":"refs/heads/master"},"f4d1a637cdd09032ecad22685d557720a5cfc826":{"kind":"REWORK","_number":4,"created":"2025-08-26 20:06:38.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/952624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/952624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/952624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/952624/4"}}},"commit":{"parents":[{"commit":"af474fb8330e8fc389ea23a57dd89fb9e8e5937f","subject":"Rework the test launcher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af474fb8330e8fc389ea23a57dd89fb9e8e5937f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-13 22:49:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 20:06:07.000000000","tz":120},"subject":"Remove eventlet and start using threading oslo.service","message":"Remove eventlet and start using threading oslo.service\n\nSince oslo.service merged the threading backend, it\u0027s safe now to move\nmistral forward and completely get rid of eventlet.\n\nChange-Id: Ie02765348a0d892648fad2b09348c57f1ac0108c\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f4d1a637cdd09032ecad22685d557720a5cfc826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f4d1a637cdd09032ecad22685d557720a5cfc826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
