)]}'
{"id":"openstack%2Fmistral~953487","triplet_id":"openstack%2Fmistral~master~I7bc307bc8ded3bf9972e44ed4bab2569c0786f3f","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-04 08:44:22.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I7bc307bc8ded3bf9972e44ed4bab2569c0786f3f","subject":"Rework the test launcher","status":"MERGED","created":"2025-06-26 21:52:46.000000000","updated":"2025-09-04 09:30:31.000000000","submitted":"2025-09-04 09:29:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953487-eventlet-removal","meta_rev_id":"29c9ae8e77eef68af68c0f962a33bc7453e23e28","_number":953487,"virtual_id_number":953487,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 09:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 08:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 08:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 22:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6747047cbcca21c469b7035b72b56221620df4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-26 21:52:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0416d7cc270f7f21ce2ffa526a3953177239324a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 22:39:20.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99920b014dc94d1085fdf8c23525d794\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04ad1b177644daca37eae625a4acfd9 : FAILURE in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ae1443444dd48bfbdc23e0dbb7685db : SUCCESS in 11m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e865c318698e4f7cb2e1c2820448b3cf : SUCCESS in 11m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f3121e08b71468b8416cd15b9953215 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94238eca7d4e413fbd28478d6b5bc073 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dc207b15f874f758328ecc4b1fa4b7f : SUCCESS in 12m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7325aeb11e8c4646b25cd7816ab54ee9 : SUCCESS in 44m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8e5ce3a37a04c9f9182e6ac54875616 : SUCCESS in 38m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9afebaddad354a28a886c75e64328aba : SUCCESS in 17m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d5479a0ba22745e8aab82e8aaafcb3c8 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f843571f6aeb6762fe31bb03f3aa8739030a4525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-26 20:06:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e079eb7906c34c443df171db83c53994cf151007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 20:31:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c04f7b4f808e4be99b4b9589b52cd423\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8d628dc91741ed9f981ec795493fd9 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/967fce28e6214aada864f514640d59e8 : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2779004015b43a48a9e7ddab161589d : SUCCESS in 10m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dadd873fca1e4025991dca3afba5144e : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/745d5a37259c49a08d2446e1fbf5f156 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132330ea5431410ba2f08077e432d7f5 : SUCCESS in 15m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/253c5408dc864a9e89f6c8ce9826c4e9 : SUCCESS in 16m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fee2c43ce51b4c1db1947d6329cd06f7 : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9c7dabc90fbd6ceb0db89efedafbac32563a92a3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 08:44:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62ab6e8df72a668042f658a3e4e7cad2eef192c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:44:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a615be76a250741d4e3e3401ba57aaca44461b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:29:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1acebdfddcac481ab6cf916b84197249\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ebf8803b0041eba1460f3c1ecc5189 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cc88607433b4d038e298192a87736d1 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a668ae5fe45459b97bf3875131879fc : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86b3b561c5142418c9d086dc02c52ef : SUCCESS in 8m 37s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/54f5cabeb68f421ca7c423752ae411f1 : SUCCESS in 16m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27f8e17c73a647139c82182d6859633d : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":2},{"id":"70d7353520db55c74184da62ddec98e66bd81ed7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:29:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"29c9ae8e77eef68af68c0f962a33bc7453e23e28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:30:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472981fcc53b494a863ca326a769edcd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8b180d1c2945c7a49edfc8e2a46e27 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af474fb8330e8fc389ea23a57dd89fb9e8e5937f","revisions":{"7dec72648d974f67363f2a098698f9d82660d639":{"kind":"REWORK","_number":1,"created":"2025-06-26 21:52:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/87/953487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/953487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/953487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/953487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/953487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/953487/1"}}},"commit":{"parents":[{"commit":"17f58d52d03b6a76761c3f36e5bc04d7cc840c6f","subject":"Merge \"Remove Jammy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/17f58d52d03b6a76761c3f36e5bc04d7cc840c6f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-26 21:14:53.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-26 21:16:38.000000000","tz":120},"subject":"Rework the test launcher","message":"Rework the test launcher\n\nSince we are going to use new oslo backend, some tests are going to fail\ndue to this.\nLet\u0027s rework the test to mock quite early the process so we don\u0027t have\nto actually start a real thread for tests.\n\nChange-Id: I7bc307bc8ded3bf9972e44ed4bab2569c0786f3f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7dec72648d974f67363f2a098698f9d82660d639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7dec72648d974f67363f2a098698f9d82660d639"}]},"branch":"refs/heads/master"},"af474fb8330e8fc389ea23a57dd89fb9e8e5937f":{"kind":"REWORK","_number":2,"created":"2025-08-26 20:06:38.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/87/953487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/953487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/953487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/953487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/953487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/953487/2"}}},"commit":{"parents":[{"commit":"db8734c053ddc91715e46517ba403a5bb7c74dc1","subject":"Fix logging format in action heartbeat startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/db8734c053ddc91715e46517ba403a5bb7c74dc1"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-06-26 21:14:53.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 20:06:07.000000000","tz":120},"subject":"Rework the test launcher","message":"Rework the test launcher\n\nSince we are going to use new oslo backend, some tests are going to fail\ndue to this.\nLet\u0027s rework the test to mock quite early the process so we don\u0027t have\nto actually start a real thread for tests.\n\nChange-Id: I7bc307bc8ded3bf9972e44ed4bab2569c0786f3f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af474fb8330e8fc389ea23a57dd89fb9e8e5937f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af474fb8330e8fc389ea23a57dd89fb9e8e5937f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
