)]}'
{"id":"openstack%2Fmistral~953502","triplet_id":"openstack%2Fmistral~master~I933c5b5b05085548ad2c4f79727bd15e340c5b62","project":"openstack/mistral","branch":"master","topic":"remove-USE_PYTHON3","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-06-27 05:55:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I933c5b5b05085548ad2c4f79727bd15e340c5b62","subject":"Zuul: do not use USE_PYTHON3","status":"MERGED","created":"2025-06-27 00:40:26.000000000","updated":"2025-07-06 21:56:58.000000000","submitted":"2025-06-27 05:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953502-remove-USE_PYTHON3","meta_rev_id":"634394a7ca3ebd591f2a642dbb3017341a761dff","_number":953502,"virtual_id_number":953502,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 05:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-06 21:56:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-06-27 05:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-06-27 05:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 01:28:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 05:15:49.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-07-06 21:56:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"8328919836d71caba25b415fbb463cff94c7b07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-27 00:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec77ce33ce629ce24cbb7d92f3c2eee2b56798a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 01:28:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10706ce729b34e3cbb3ad7efd9978b5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/689328926b01493d90fef4e96a4e83d9 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd2316ce964349de8257feea35aa0df8 : SUCCESS in 9m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/827bf799b9a14582953477ab03232a51 : SUCCESS in 12m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75342581abbb4d29a854cffcc9732b5b : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8805ab521a434f5e871a9cdd822fc335 : SUCCESS in 10m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6715c2af763d4fd198a8c66cefe8b78b : SUCCESS in 16m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dff373f227054667b35c24a6965ca145 : SUCCESS in 43m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9985d2db36ba480dabc36a86e44c65de : SUCCESS in 19m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/04cf11e2f2d244c99d33beefdd068542 : SUCCESS in 27m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f4e3aac62c377ccd2e2546120e22dfce35fe486a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-27 05:15:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"068db37fb49a3b296852be12afd5807310d5757a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 05:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c9c8b6e46c452823e8cd5e40377065f41e72163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 05:55:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e8c4cf28e74eb0b3ceb4166052630a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2e7b48993841e2b175415cc04db7f5 : SUCCESS in 2m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d087d0b01f147c3ab42185bec8e9c94 : SUCCESS in 9m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32b4170eadac4c33a27c2f07a91c8056 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132bb09f3e8e457e9ee4b40b25d3c637 : SUCCESS in 4m 13s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4fc747d78cdc442e8f42d5a3451532e5 : SUCCESS in 27m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a7a926ac7a548b2b9d6f104d2bde573 : SUCCESS in 36m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8848c18e361e44b99d8768ed21e8f718 : SUCCESS in 35m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8361f0cbd9f347259a6e857d0f74419a : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7b3478afbdf01a2624db454aa64e2421a393f2e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 05:55:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e10b6a4f81009134c737ddea2c522d878acbb8f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 05:57:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a02b8e25b24169b6929d55a5041042\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16094fb71dbd4155b55fe9b0ca6cd733 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"634394a7ca3ebd591f2a642dbb3017341a761dff","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 21:56:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ed89e8f51313628224be0a42d3f71a8b2186dc3","revisions":{"6ed89e8f51313628224be0a42d3f71a8b2186dc3":{"kind":"REWORK","_number":1,"created":"2025-06-27 00:40:26.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/02/953502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/953502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/953502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/953502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/953502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/953502/1"}}},"commit":{"parents":[{"commit":"17f58d52d03b6a76761c3f36e5bc04d7cc840c6f","subject":"Merge \"Remove Jammy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/17f58d52d03b6a76761c3f36e5bc04d7cc840c6f"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:40:11.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:40:24.000000000","tz":120},"subject":"Zuul: do not use USE_PYTHON3","message":"Zuul: do not use USE_PYTHON3\n\nDevstack has removed the USE_PYTHON3 variable[1][2] and now always uses\nPython 3.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/920658\n[2] Commit 5412dbfe7b797149f1f68100de8003b1876398fe\n\nChange-Id: I933c5b5b05085548ad2c4f79727bd15e340c5b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ed89e8f51313628224be0a42d3f71a8b2186dc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ed89e8f51313628224be0a42d3f71a8b2186dc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
