)]}'
{"id":"openstack%2Fmistral~954953","triplet_id":"openstack%2Fmistral~master~I33f9515fe5f0b45a595f345e36abe024cfa1645f","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-17 16:06:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I33f9515fe5f0b45a595f345e36abe024cfa1645f","subject":"devstack: Remove coordination","status":"MERGED","created":"2025-07-14 14:19:55.000000000","updated":"2025-07-17 17:12:33.000000000","submitted":"2025-07-17 17:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954953","meta_rev_id":"edf484e36e346ebeee97b35c7d6ecc48d9478d24","_number":954953,"virtual_id_number":954953,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-17 17:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-17 12:13:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-07-17 12:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-07-17 12:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-14 15:03:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 12:10:26.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-07-17 12:13:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"3be7f433a2849a05446a55ddf98eb215862f496f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 14:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b699a09a1343a51edd21a5af5ee82c3dfa2530c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 15:03:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c1b872e8417455a922c2bdf14bc2760\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a8e47214d64dc69e08c620f085c5b4 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd08e04f208648fbb3445eba53262f6f : SUCCESS in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21f4c51be0274f10b313fd06d1b83961 : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68e31767a0e44f7694175244d27d5189 : SUCCESS in 15m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba3d17b72084361b4941cca43b5405b : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae6820c1b2dc4611ae78e23fdc986c4a : FAILURE in 17m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a5f2fc99c664e5ebc0909990461e0b5 : SUCCESS in 25m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38ef4a704a264b9083ebc9beaede58aa : SUCCESS in 30m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f3918c470364ba4b872ac6e7d4a2fa5 : SUCCESS in 26m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/31652348550742818742aace808b36c9 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4d0731f9cdac82322031f802d40eb1d1509d0242","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-07-17 12:10:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14d31a5b3ef55770ea93b52036a3185882d2defc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:10:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"298f9ed5c64c761dc0c43edbdb4b8f77deeba207","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 12:13:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d17ac33519d249c88a367173b9afaf384ecebdb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 13:13:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0e977c287e54a9ca80ff31f3c025edc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceaafcf79e1746a7abcd1a4ce7ed3853 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6ba760d607543bd894e2222c52b0483 : SUCCESS in 9m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/094cf8391ead484ab1f1f959ec058cba : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4491d18602544eb5b968558fbdcc6477 : SUCCESS in 7m 53s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/990854288db5489e9535817c2bb94a9e : SUCCESS in 39m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8835b112b5d946c8b4f18df01d1f7566 : SUCCESS in 40m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3ae2b65064540f08d416791eb82c797 : FAILURE in 49m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/31a97d6579b1418ab8b06ad83581d252 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e36a32af4daaf284462a93dbdd40c8f4115f76c6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 13:21:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04533518d13313350e1504f42b73fbd801848b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 14:19:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8f5ac49499c4523b29d5667862d49f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7781ca42f5064e74aac56ee7fd9ef903 : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbe1ce6063f34e3c97721379be814721 : SUCCESS in 14m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6da3439747b048db856cbcf1edb06209 : SUCCESS in 9m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d488497c94034e6bb325cb390ef5e59a : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a73ec7bd5d747a6a82370928f2636c7 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99619d04ef1d4278a02a9ec8922fe3f5 : FAILURE in 18m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fae08f1c9e3b4c62866d30294a0d8260 : FAILURE in 38m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0859dc874e74f60b7e9f2634ad1cfff : SUCCESS in 49m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6833d852d05452facdbd4f12a7bda33 : FAILURE in 37m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/22eee91841164f068ad707a6a13f9be1 : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f44ae1b45aea330d0c833d740ed0c366a0b8caaf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-17 14:40:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e21e856784133875ea00bf3e6076340e0bdf0bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 15:22:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fdd66be514f4490bb710557e4484798\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f64529a6fc4a1cadeb569c345d0645 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50914324349946a38083aebe74f31c59 : SUCCESS in 13m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdf6ad7fb54a43f3a631507562d4497a : SUCCESS in 14m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8fcc802149948a493044f6490e379bb : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae3007155954da3ac5d4b664695af64 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/129c3ce735104833a726cedd0ebab366 : FAILURE in 17m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5225078e071b4609bdc8618233da5bbd : SUCCESS in 17m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60e57c78b8104a29a70200c2afd52254 : SUCCESS in 26m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/162b5fb301e849b0ac465f718148e991 : FAILURE in 40m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a6c401049f9643f4bcb4cb7d29eee462 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":1},{"id":"212e10a75e0cccb200093a9dbd94885653347772","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-17 16:06:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae0771e9659cda55c51bf94e916ec37326c6f207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 16:43:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b0c08b1bd6347afaf67cd27e5dcefb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfc3f200b6349e9afa6967457d7d9f3 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac710385007e4953b9103081dc52b565 : SUCCESS in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adbd818a0c13406b8b0f20f166fe211f : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edc3b5de972440528c3664ecc8dc4f16 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e53ad05b18e433da50893ae2acb719f : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5190dd8771204012996bb87f1407ed85 : FAILURE in 18m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0a34ac8e824f4800b57667601700237e : SUCCESS in 17m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f420c0f92504b67b20eb231dfcab5b8 : SUCCESS in 25m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e017d0236bb2443183d3a7dea1e64605 : SUCCESS in 17m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ef645981286648f2b1e3f94385d24821 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3dd1b71000593762a984507cdf9236777e91f93a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 16:43:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9783e32f2786285858f6bae84d570399d398c54e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 17:11:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b2e16d1450411d901932ea72e68b45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0a9093b533472b96d8832d28afc4a9 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6218ca795b540948d2cad1d2f7b1665 : SUCCESS in 10m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/633cf0ae6f0f4987806c29e14dc59733 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d05d3d992a6b4c2c8a4abbcad69331ac : SUCCESS in 7m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0c032db9ad2c4c988bee4544fbd5c0d9 : SUCCESS in 26m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d3781d0cf1e46dc83215a24248750c0 : SUCCESS in 27m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e136635a0e6e434587e96b7b811e5f75 : SUCCESS in 25m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f9d01e3cbfee4993ad3ad8dda8917942 : SUCCESS in 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e38407729276cf274d49e6ee1f902f71c9074f73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 17:11:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"edf484e36e346ebeee97b35c7d6ecc48d9478d24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 17:12:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189f42b84552466cba04fd50f2007cb5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4947a91e4d44e8caf5db83853349711 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e50c6c854d4ca0d6b8ec657d8a93c17d16b750d","revisions":{"4e50c6c854d4ca0d6b8ec657d8a93c17d16b750d":{"kind":"REWORK","_number":1,"created":"2025-07-14 14:19:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/954953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/954953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/954953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/954953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/954953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/954953/1"}}},"commit":{"parents":[{"commit":"6ed89e8f51313628224be0a42d3f71a8b2186dc3","subject":"Zuul: do not use USE_PYTHON3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ed89e8f51313628224be0a42d3f71a8b2186dc3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 14:19:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 14:19:46.000000000","tz":540},"subject":"devstack: Remove coordination","message":"devstack: Remove coordination\n\nThe actual implementation was already removed by [1].\n\n[1] 6a2e451e21032ae983729bbf4ade0c5d82c0929d\n\nChange-Id: I33f9515fe5f0b45a595f345e36abe024cfa1645f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e50c6c854d4ca0d6b8ec657d8a93c17d16b750d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e50c6c854d4ca0d6b8ec657d8a93c17d16b750d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
