)]}'
{"id":"openstack%2Fmistral~954957","triplet_id":"openstack%2Fmistral~master~Ieaf1fa00f7425c9fef9ad0f5d51cce587d92b7b5","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-17 21:29:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieaf1fa00f7425c9fef9ad0f5d51cce587d92b7b5","subject":"devstack: Remove rpc_implementation","status":"MERGED","created":"2025-07-14 15:11:08.000000000","updated":"2025-07-17 21:29:56.000000000","submitted":"2025-07-17 21:29:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954957","meta_rev_id":"47532973fd1d62fc237467c5a3108b5a336511de","_number":954957,"virtual_id_number":954957,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-17 21:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-17 12:13:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-07-17 12:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-07-17 12:10:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-14 15:55:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 12:10:30.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-07-17 12:13:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"6c25fe99234927f633ee030a5efee18f2552d20c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 15:11:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b8820f6a06816ce858de3cf3a0d5547874889f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 15:55:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa1315987b784dbb90a5fc4563259fb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2982feae130f4337b73684ffe1148bb1 : SUCCESS in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24a63ebc46d54f829b238de317492914 : SUCCESS in 10m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/813cc300aa884cc390fc0f4b7fad57f1 : SUCCESS in 9m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4124b59c20c143a4a3e0d53f4af23253 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd57a2d8748b4ae584028ecb4c391a22 : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4cde80b6c894d83a0fae02462fdb629 : FAILURE in 16m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b594a0388f1142eda1085eb90ca29b05 : SUCCESS in 17m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d34481b50b4439f9a1b3fbbe9d825ec : FAILURE in 39m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06128341832f40c9bda71054df641012 : SUCCESS in 40m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b337378d2bc1410ebb32676012e2b095 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6acef53c5ee7e8f07ca06b78213e4bc8a2873143","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-15 01:13:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eb11219b61d46918194d0d71c80a754c504dac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 02:03:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da43782e6bda4af0ad87f459d6cbf20f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346b9b0122c9488c9163c39d0aedf514 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39cbf6b6ddf34a32a63fb057d7d4d8d4 : SUCCESS in 12m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d1e3f14c1e5425890f03ffe3bc3fec3 : SUCCESS in 9m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55668592511645bba9303aa5752b6417 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b895c90b714e55b873cf52026c3ddf : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cbb91dcf90449b681d1e9223a96f3f9 : SUCCESS in 16m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b7f7cdb2f1ed479c87ce6ee2e0e63632 : SUCCESS in 30m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4233feef866142c9b664aeb1aecf9740 : SUCCESS in 43m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/370b072ab4f64bca9c1d6230ccc3eb58 : SUCCESS in 37m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/17e741e5b52f4400811ce7088d4fde92 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9a9181d2b8ba1a9a9b92fac3b3b5e1f5bb02a659","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-07-17 12:10:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea4affff48a105e181cf855641942191164d0c1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:10:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5800ca3b554631739f6c42b96572897ef2df18ac","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 12:13:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6660e714b0d44ea9315c6a0cebdcaaac04e85ed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 13:13:49.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/954953 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"71ae375964679d12b4349b5e743fed410feb4de9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 17:12:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"029dcb9f45d3cbb7dd5147fe7a729751ad5da48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 17:51:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d22097616d4067b80452f5662c7de2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5188ff179aa4ff28b8401ae894ef3c5 : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17cd0859b6ab49fd859c5d1a5651a265 : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97df66db692e42fab37640dd2f308d51 : SUCCESS in 11m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e458947d8c9244c397aa4f7d68829378 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c73e2297cbb5446f99f15610070bf238 : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/023cd56f8def4604b7786c0493c9685e : SUCCESS in 12m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4129212b1e064c698009b0df8670d897 : SUCCESS in 36m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8099cec884d4bba9f41192a31573e6d : SUCCESS in 26m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4d71333e0ee4181ae312e906e3f98aa : SUCCESS in 33m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ea4a234f7e9c4f639b277e0a6ab520b0 : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8bf819d3533cc41a853662aa5ce53639ad9ae5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 17:51:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f740d57e8475121e7ac7df5dd930284db6ccd61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 18:42:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ff6bbb7191949f2b07d81443a48974b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1eac998c7c54e44a31b2de0c7b02bbc : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bad112e241b94aa48e5c12c6ae048620 : FAILURE in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c67e2cee5264b4aa21f1673a4350bfa : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0f2fc6243244f799839308596cd1c7 : SUCCESS in 9m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d38d5ede8c6c474aafd2940f27869da4 : SUCCESS in 30m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7f4c34f66d74cca97f729f14b0508e4 : SUCCESS in 46m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65876168d9694722bb3038a9c93b586d : SUCCESS in 37m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2e4eb35736484a408a81fd1cfa7159c5 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ed80008b40b9ee3a0bf75f93c9e28da600960502","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 19:50:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c098551740bfcd0eba18be9ff978dadf825ab64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 20:44:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32a5b2b2c6741f18f95be5de3cebd68\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dd71cc5eb54b2b935af9e4c4270942 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be2e75bf5a7546eaa28a555146728ba2 : SUCCESS in 14m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/550e5366508a43bdabeffbdd67468fcd : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a73edf4387274e3b9d7b9af2593bbaca : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c2e448467b49e09d0af75992e5a8e6 : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e4a1f3e5df8498782912d0d3d63c052 : FAILURE in 21m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/55819b77ee974ceaa12762a956b86572 : SUCCESS in 39m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/193c0593e0f14641aaf9cf867e051ad5 : SUCCESS in 42m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5973b9536f4e4fb6865165a3b1f94b6f : SUCCESS in 40m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/04d3c983c21d4e5bbd7e7e8f158d81b1 : SUCCESS in 44m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7689af1f9cece44a830c0aa89a0a112a911828f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 20:45:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"923d2387e6a01259b7dbee0f7f9b96c7b0bdb309","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 21:29:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79df5e025fd0401ea0e4ed9cebb9b4c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993171e84bec492998f2a5402ea72463 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3889c26ae3b4758933967a5dddec6b2 : SUCCESS in 13m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e5a1902c0544cd0b9cdaafb52c02722 : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33adb5c325dc4e4a897ad240cfe32142 : SUCCESS in 8m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a5f51706964f44088b1dce6f9416d273 : SUCCESS in 38m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d87cd7867b1471794d22a0ad0eff32a : SUCCESS in 36m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69758d366ab94562a0f99f7d7092ff32 : SUCCESS in 39m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5331e691947f45eeab11f489f735a234 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"35e3c3e153e6e5eeec9da3710be047cb6b94502a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 21:29:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47532973fd1d62fc237467c5a3108b5a336511de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 21:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf2c86af290a4dbb97db27d244c32370\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f290d62f8d43abb869beafcbe369ed : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e08a60977b70d8ecca69e293a07da7fa61ffc002","revisions":{"e08a60977b70d8ecca69e293a07da7fa61ffc002":{"kind":"REWORK","_number":1,"created":"2025-07-14 15:11:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/954957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/57/954957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/954957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/954957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/954957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/57/954957/1"}}},"commit":{"parents":[{"commit":"4e50c6c854d4ca0d6b8ec657d8a93c17d16b750d","subject":"devstack: Remove coordination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e50c6c854d4ca0d6b8ec657d8a93c17d16b750d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 15:10:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 15:10:21.000000000","tz":540},"subject":"devstack: Remove rpc_implementation","message":"devstack: Remove rpc_implementation\n\nThe option was removed by 715fb3db7667205dbcdf361e4ccf51f585fa19dc .\n\nChange-Id: Ieaf1fa00f7425c9fef9ad0f5d51cce587d92b7b5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e08a60977b70d8ecca69e293a07da7fa61ffc002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e08a60977b70d8ecca69e293a07da7fa61ffc002"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
