)]}'
{"id":"openstack%2Fmistral~956346","triplet_id":"openstack%2Fmistral~master~I7633714dfc640ec7f18a963538e33df55ac20660","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-02 20:11:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7633714dfc640ec7f18a963538e33df55ac20660","subject":"Parse server option while parsing its input","status":"MERGED","created":"2025-08-01 09:26:25.000000000","updated":"2025-08-02 20:12:30.000000000","submitted":"2025-08-02 20:11:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956346","meta_rev_id":"db454049613f0ca39c74742fe67580f3a323a50a","_number":956346,"virtual_id_number":956346,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-02 20:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-02 19:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-02 19:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-01 09:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-02 19:27:39.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"8b8c0c87898fcdc73f29ed121332ca25bec9d5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-01 09:26:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eab3365b365094b2e3df6e3f08295c63baf4b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 09:58:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e95af5534853413ebd3ff0fcf748e5d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4630d87554f34fa2836b91d19b1d3284 : FAILURE in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a734aa4dc944467eb3023063796e2393 : FAILURE in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c40a607c168494bb21fe440c2e67647 : FAILURE in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e1757a31978456a87b5cdb53045bdfd : FAILURE in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2321ed8d6730456e8202e16dd1b63c24 : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52541d7595c4bb296a78173a28184ad : FAILURE in 4m 34s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/629755c279ae4e50bd55398bf69f1808 : FAILURE in 15m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c495ef56b3d84dd8b078315e6cbe4123 : FAILURE in 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0ceddbc5bb696c12e2799af63539374fd295296b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-02 12:55:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"976728fe0cfc2f364d9e8c62e068cca8b01ade4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 13:35:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950f779220ad437a91cd0e564c8c2b2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e899ae1bf04ff68ce6ff93bfab094e : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fb18e2333ab41a5a97d2db4690a17df : SUCCESS in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aeca89f9f9f34ba5890d9ae26832109c : SUCCESS in 11m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3789a39a510e421dad3e99beb28b1774 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69679a5055549088acf484fc825c738 : SUCCESS in 9m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4579f7ee156b4af6bf7984700a99b9d3 : SUCCESS in 12m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/62406cf5a48247deba3e2d0220148b24 : SUCCESS in 35m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfaee7cdd6cf44468115139755b3c19b : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e8a841a04c2df9f3232dec48bd676683f3bee659","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-02 19:27:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dbc89a7ca112dfff51f2255141a499a9e532d7c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 19:28:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da604337f8b6cc80875faf9e11566884fd645c4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 20:11:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce917bca047472ea4812cc119696a74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a8bb2d58a04e809f626235f871a443 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/935d7d9c6efc43538a7d7a9d59b839a0 : SUCCESS in 11m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/361edd0d083d4ff58b883b6307c37eaf : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becc3a1fb2794882b54e836b793246e9 : SUCCESS in 8m 05s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5993a0468ca2403bb277e1ce7ec8c6f2 : SUCCESS in 29m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8260406bb41e48478a0dd377e7816049 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"594733724a1ed48512e1a35eb4b32897adf2d641","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 20:11:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"db454049613f0ca39c74742fe67580f3a323a50a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 20:12:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/418bd0864c3c4bbbac23c18a51507e88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c72f62755aa54502a805b687c92c1048 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"677f4fba18d8921e8616e44a0db843869b7c828b","revisions":{"41c18bab425ff792f22d93524d087861557dd699":{"kind":"REWORK","_number":1,"created":"2025-08-01 09:26:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/956346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/46/956346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/46/956346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/46/956346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/46/956346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/46/956346/1"}}},"commit":{"parents":[{"commit":"6df487336ef33cc77de71c06e0b6dbf82285d5dc","subject":"Use systemctl command to manage service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6df487336ef33cc77de71c06e0b6dbf82285d5dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 09:25:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 09:26:06.000000000","tz":540},"subject":"Parse server option while parsing its input","message":"Parse server option while parsing its input\n\n... to give more early and clear feedback to users.\n\nChange-Id: I7633714dfc640ec7f18a963538e33df55ac20660\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41c18bab425ff792f22d93524d087861557dd699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41c18bab425ff792f22d93524d087861557dd699"}]},"branch":"refs/heads/master"},"677f4fba18d8921e8616e44a0db843869b7c828b":{"kind":"REWORK","_number":2,"created":"2025-08-02 12:55:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/956346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/46/956346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/46/956346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/46/956346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/46/956346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/46/956346/2"}}},"commit":{"parents":[{"commit":"6df487336ef33cc77de71c06e0b6dbf82285d5dc","subject":"Use systemctl command to manage service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6df487336ef33cc77de71c06e0b6dbf82285d5dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 09:25:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 12:54:57.000000000","tz":540},"subject":"Parse server option while parsing its input","message":"Parse server option while parsing its input\n\n... to give more early and clear feedback to users.\n\nChange-Id: I7633714dfc640ec7f18a963538e33df55ac20660\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/677f4fba18d8921e8616e44a0db843869b7c828b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/677f4fba18d8921e8616e44a0db843869b7c828b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
