)]}'
{"id":"openstack%2Fmistral~958595","triplet_id":"openstack%2Fmistral~master~I55172351d1768bdd34fab50bac228928ce7b1765","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-04 08:48:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I55172351d1768bdd34fab50bac228928ce7b1765","subject":"Replace eventlet wsgi by cheroot","status":"MERGED","created":"2025-08-26 22:14:07.000000000","updated":"2025-09-04 09:41:40.000000000","submitted":"2025-09-04 09:41:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958595-eventlet-removal","meta_rev_id":"dacdb65e333806a11a94e3652884facf7a86b25f","_number":958595,"virtual_id_number":958595,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 09:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 08:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 08:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 22:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de1345a32785a7e2ff2677d14b3c81e37d2a089a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-26 22:14:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4f5a7c65e5b07ab97b134c422c5639dbd5805c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 22:58:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26929c05ea544c4e84e42d3c577e4d9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012c943dcb2e42f78a335006a2699da6 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc2f8b294183499bb7b123bcef589334 : SUCCESS in 10m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/886290ac3a584e1aaaf070ecbd1af720 : SUCCESS in 10m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e6b6281f9b34d11898665ac01c58b8f : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efae29b17bea439a9126505032d6b0e4 : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca62438695c7429c8a4b31dc3dc60656 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4bb444b0e9a4004b8e84d2d8278fe72 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57343fb23d244b0983d87c82015bf769 : SUCCESS in 15m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6e4f4df3ba474790b07179e8ebfd51a7 : FAILURE in 17m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d47cf001ca4e4725b15bf7189dd12048 : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5bf3d47467a34673e153684a2dd25d95b3cd2be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-27 18:05:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e407603a04d6d84bd13932349f7a321f0e5bb4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 18:36:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3acbf48184d84d909f333509921cfe7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad642b69063b4e26acf38c0bde95a0a3 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01b391a0b9b84bddbc3847592ce96cb3 : FAILURE in 25m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b11ffb740e742f299fa57395bfe269c : FAILURE in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ecd63da606140699aa350fa69eae651 : FAILURE in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bfed86d55c94901a555761355fa88f4 : SUCCESS in 7m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae5133a681914bb7908776f4cfd01068 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1859277bb8e433492f0f0cee37e66ed : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8eed88163d3a46a3820af94ef3ac5841 : FAILURE in 4m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/60c7dd6905b446249505b2d397f5f9d6 : FAILURE in 18m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a7e3fba6f2a4c25909fe282f1426772 : FAILURE in 11m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c95a1e390bba2e82ef870dfdb9ff5076714fb196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-27 21:53:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"23e5bced851967f6ff5b2a1f31753e77716d0b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 22:47:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdc8934766274e6d9bb113ce1dd2932f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498db4fc4bb5492baf1b86b8cf36bd71 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f6e4b9a15794239b570e5dfb93a5057 : TIMED_OUT in 41m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8c5c7497bed45809e9f74f96beece69 : TIMED_OUT in 41m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bde1eaf43364070984ac6ab3fc52ace : TIMED_OUT in 40m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07aea252c1b4221aacbaad138a3215c : SUCCESS in 11m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f713bd76a9e74835832c83ce5bb0f6d3 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c61d9e6f5ed643e897272e96137560d1 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd89d6d850af4650a63368f1c02e1513 : TIMED_OUT in 51m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/215f5baa8ce04b8cb8a0c463e85ae2eb : FAILURE in 14m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da013d394c0431e8b00ad34609dbe65 : FAILURE in 19m 57s","accounts_in_message":[],"_revision_number":3},{"id":"c730d3eff06afd7b1a3e306238676c934780a7f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-28 11:35:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"85a8f4acb3c3812603efd3f94db95f0d20458492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:29:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6814f6c8c3ad4e7d953f82d0db718ce8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5299cf8696dc4d82be5277b5035b0f55 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13ab8cbdc37e4ad5bad7218d62138e23 : TIMED_OUT in 41m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6dbb4e48259464aa61670625495cde0 : TIMED_OUT in 41m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baeb3178b7e04a148f9da7a5c2d87530 : TIMED_OUT in 41m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20531a27c2164b638c831d4926facc09 : SUCCESS in 7m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d9aa14d6b1a445f9704c4fb70bfc0b2 : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9085d4d4e164e5bbe4a2098817cdaa9 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/345d2865b215458da9423f90bfdb4a3d : TIMED_OUT in 51m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d0a031d3fba543489d0352ece3d00d4e : SUCCESS in 24m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dd5bf87d2ee43398d204b344eb5e251 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":4},{"id":"408c7dfba8035274aad4dd59a541a7a0ed1917e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 19:12:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e0c43df2cecfb53337255e391c3654ef80edad94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 20:14:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/005ac6d9e60945fa9011c53349320bc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe1d57469ff4757acb3694d6d700a72 : FAILURE in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ff19cbc2474a2f81a103a7f332f8f3 : TIMED_OUT in 40m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0bc3db9af844697beb696678e10e6e3 : TIMED_OUT in 40m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29acbedeecb441b5abed2b6f7f156bab : TIMED_OUT in 40m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558efbb9e24445ce862a38574398edd7 : SUCCESS in 9m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d41cb958b899431e8f6cf869301f7835 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a624e1f5617f42109b54dbbe8143bcce : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce4ea16eb9bd46e7b1647b19bb5c2289 : TIMED_OUT in 51m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/527b56a50978415eb324ffdae621924f : SUCCESS in 37m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01d0749c88174ddfb2db99a080ece611 : SUCCESS in 22m 09s\n\nWarning:\n  Comments left for invalid file mistral/db/v2/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":5},{"id":"b2742a9828464d7c21deefb2c291461b5b5f4ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 22:50:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f57a50d51a9f99d3fd2ee5be554fc7d44ca34f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:01:32.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"9df18e519d3ab1ee32bda20cd5f159c69d4b8120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:06:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9093ab15c07468265c7899055914a07c351ea137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 23:59:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/739c82eb13244d46823c209bdc628199\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c962250f0e94c63a1695c6bb8a32299 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb8fba2134aa42449ffcefb798ea86df : TIMED_OUT in 40m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/656dae0b14c94561b73546097d696aa7 : TIMED_OUT in 40m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6784044c55e74784887eec4fe85a31cc : TIMED_OUT in 40m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c40019279d24fde9279649301f249d6 : SUCCESS in 9m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cdc2aea01f04e0f8089a50875c3819d : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ea815a59c014737aaf5bffbf6b71e32 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14065a4e30f24e7b8d43f87949b6994c : TIMED_OUT in 51m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a9f5b5bf8ba54ef0a97701d1bbdd6e45 : SUCCESS in 38m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8850c8951b95485a8aef67eeb342b691 : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":8},{"id":"55c8c0937cef5f31ca45e7c2c6406647e4b997b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 19:56:46.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"0e6a86113c1edc2a544d345011dc44c5d5e9d284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 20:16:02.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"3d0b3f635f4f076695f01a46c44996d84b7b8a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 20:44:08.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c7f27576aa645f898a7a2c955f1faf5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6999264704474a1fb2c0349dd44c61e7 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36111b382fdf4fb78fa67d288838cfa7 : FAILURE in 11m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76c8741db0a74f1ba27ab6ec7ba04636 : FAILURE in 9m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fa62c9abdd6482b8d47d7500932bfaf : FAILURE in 13m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51de93dc181f48c19a0c5246fc37e1e0 : SUCCESS in 5m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec46427e3a964ecfb7fefe65efa2e3c5 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5acfa555ca94becb4b966a44c8a377e : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c33ea775906c4b909b369b3d0c051ba2 : FAILURE in 17m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a6347e3de7af4b749e7c2adb34da3976 : SUCCESS in 26m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d10eca508cd14361bcaf37abc7baaecf : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":10},{"id":"4f88836126e31055b1728b4bc51ad396ebb4b42b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 22:02:17.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"a23f6a632f26089c2cf8ea3baedbe9f2dfaf9873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 22:48:42.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97a3bde0d1434226b5577b4271027374\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0dd188d43eb424e9a74dab97dc5d808 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86e4f22ab2f64f09b323f591661064c3 : FAILURE in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2eb71059b8b48b5be40239864628436 : FAILURE in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/281880087b07441ca908fb5a5df273b7 : FAILURE in 14m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb055b5d33cd4ddcb6b4e59182d9a224 : SUCCESS in 7m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff9b2ed0b3d04e1c9ba702756f302da9 : SUCCESS in 2m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7445d3dc5aed4537bd6533a904333de3 : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a50c321c3bcb4c3baac9c961295259c0 : FAILURE in 16m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/33908bf68a344cd98a184e228f2fb36c : SUCCESS in 16m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48db5a3f3a6f4549bbf2158826537306 : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":11},{"id":"fcb942dc43921c18b07dce65c1516353940756ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 19:02:46.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"2b4c4439b859ddae5f44aec4c0c1bc4e82ee8b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:00:58.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dcffbdbf1d04145b124cb6790d6b5fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ccba1feb83946a18ffbbdf32bea3ad4 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0b26733e1994279bd83e838cad10884 : TIMED_OUT in 41m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d921d94317b644c4837103227225e4e4 : TIMED_OUT in 40m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44378344a3f04f34877991cad5375344 : TIMED_OUT in 41m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ece592266da460f981b952fca64e9d0 : SUCCESS in 9m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0311107c49f045a0804aa54744136cf5 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cd40776aaf540f8b3a52648effcb3ee : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ad58d7502c3433cbcc485dbac756d44 : TIMED_OUT in 51m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/efbc4f05d514487aa16d4a677092c014 : SUCCESS in 40m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cf182ee8e73489fa20f4871fe37f3be : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":12},{"id":"e6bf832eb76e04ed5e0bb2ce62304966b96a3ce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 20:43:28.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0e736d1ef37b0884b868d43c2067fcc98b3aa661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 21:00:37.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"d85520092b597340d74e6ceeec803110eea631f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:58:29.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41fcd4546afd437086fc66a4e37ba8f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e501506950b4a38a2e9864a1713b8e9 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7da0a26d83ad4d2baa570228728881c0 : TIMED_OUT in 41m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c41f5f23305547118b407418bd36880e : TIMED_OUT in 40m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08896dc596bf4951a9c5c5d63cae42f4 : TIMED_OUT in 41m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3f7d6659664955a69efbb8d73d7c3f : SUCCESS in 10m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99afc84ee607486481597507360cbc3e : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45fc1c795b674dd9ad6a5006bb7d3d58 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3cdb49dd1e24601bac3418b3ddf06ba : TIMED_OUT in 50m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d998c74ab5b44c32beb38b1bfca20075 : SUCCESS in 25m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d6243ae89bf4851aa4066aa21b369a2 : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":14},{"id":"4cf20fa8acea9cba343e5e43cc45e84993053c99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 19:13:10.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"771af3a815e36092a4533df326c40998e79ba820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:11:34.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9901a76d699747daa8a1b4251b1f04ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a7e5495a37470cbcb35f7e088d4dca : SUCCESS in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/12aec74b922e4e3d8eab15c4d09d1414 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a14e8721b1b44fab10ce8fb59904056 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ebad7938ec44e22941398955e611c8f : TIMED_OUT in 51m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4380689018aa4b468704d5e539aa9259 : SUCCESS in 38m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffe30ace2be947afa1e808e0cab2e000 : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":15},{"id":"860ce1a6b72f0b89e27795b7126da71a037d3515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 22:04:13.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"681352d51ccf7f938c24d9dbcac90e50fa2d4a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 23:03:13.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/076e0fa2601a4d62894ce2c7df57d1f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac1dcc7430748af89264fa408c5dd7e : SUCCESS in 7m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/28850ed6ad1e45e69d66866609e7c00f : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a2b3c9d37494e809db12808e71276ec : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc6dd5fd052445eea9082423971a2213 : TIMED_OUT in 51m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/35706271566a406a852cf811df0162b1 : SUCCESS in 37m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6bba0fd4dfe4a649af8d176b6dd5e27 : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":16},{"id":"b1ebd635b9dc8a024dd7fd6b27b7a3681d4f33da","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 08:48:36.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"ce67399f4635bf98ddf55e0d8f7309e85d5a91bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:49:38.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"f9a4945200f524cc8d40202964fcde00336b3d2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:40.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fdaa4933e164ec6857dee1753c02940\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb40deaec384a1da21b4911a9b42cc1 : SUCCESS in 9m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a46271aabe3c4fd2a9e1e46372df786e : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf57c89c67e44f4e904059eec0f8b689 : SUCCESS in 3m 30s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/754eee8f61bf49e1b9a4f5499eb6eb0e : SUCCESS in 41m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78e9133037ab45a29bce2a915678e698 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":16},{"id":"dacdb65e333806a11a94e3652884facf7a86b25f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"58690dfd7d54b4da4663f51979cb72a3278c31ad","revisions":{"0b2a01435c13c9a3319f20931917f29798fa1611":{"kind":"REWORK","_number":1,"created":"2025-08-26 22:14:07.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/1"}}},"commit":{"parents":[{"commit":"af474fb8330e8fc389ea23a57dd89fb9e8e5937f","subject":"Rework the test launcher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af474fb8330e8fc389ea23a57dd89fb9e8e5937f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b2a01435c13c9a3319f20931917f29798fa1611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b2a01435c13c9a3319f20931917f29798fa1611"}]},"branch":"refs/heads/master"},"d6402179d2033aa4e9073dfd66d445f33e9b2ada":{"kind":"REWORK","_number":2,"created":"2025-08-27 18:05:41.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/2"}}},"commit":{"parents":[{"commit":"af474fb8330e8fc389ea23a57dd89fb9e8e5937f","subject":"Rework the test launcher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af474fb8330e8fc389ea23a57dd89fb9e8e5937f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 18:05:33.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6402179d2033aa4e9073dfd66d445f33e9b2ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6402179d2033aa4e9073dfd66d445f33e9b2ada"}]},"branch":"refs/heads/master"},"e9d52ca794afd0b4d69e87bbfee05b33a2077078":{"kind":"REWORK","_number":3,"created":"2025-08-27 21:53:06.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/3"}}},"commit":{"parents":[{"commit":"da3eb2bb56bf31cf82b5df950bd591689bc8a60a","subject":"Stop manipulating sys.path when launching services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da3eb2bb56bf31cf82b5df950bd591689bc8a60a"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:49:13.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9d52ca794afd0b4d69e87bbfee05b33a2077078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9d52ca794afd0b4d69e87bbfee05b33a2077078"}]},"branch":"refs/heads/master"},"df549b3625d5908e99c1cd6d5b1c534929d363b1":{"kind":"REWORK","_number":4,"created":"2025-08-28 11:35:42.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/4"}}},"commit":{"parents":[{"commit":"da3eb2bb56bf31cf82b5df950bd591689bc8a60a","subject":"Stop manipulating sys.path when launching services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da3eb2bb56bf31cf82b5df950bd591689bc8a60a"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 11:34:45.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df549b3625d5908e99c1cd6d5b1c534929d363b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df549b3625d5908e99c1cd6d5b1c534929d363b1"}]},"branch":"refs/heads/master"},"6ea3654cac754947a206fc1c603a69f7773a5f90":{"kind":"REWORK","_number":5,"created":"2025-08-31 19:12:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/5"}}},"commit":{"parents":[{"commit":"13c26c35a1911b2f4b46dcf235e2c9e214762f27","subject":"Use a threading.RLock for SQL sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/13c26c35a1911b2f4b46dcf235e2c9e214762f27"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:12:10.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ea3654cac754947a206fc1c603a69f7773a5f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ea3654cac754947a206fc1c603a69f7773a5f90"}]},"branch":"refs/heads/master"},"3d476ef8108e8ad33499b7356784559a49ae2c77":{"kind":"REWORK","_number":6,"created":"2025-08-31 22:50:05.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/6"}}},"commit":{"parents":[{"commit":"d7c7f4c896e6e0d6e091fe6998cbe0c53786cfa5","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d7c7f4c896e6e0d6e091fe6998cbe0c53786cfa5"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 22:49:02.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d476ef8108e8ad33499b7356784559a49ae2c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d476ef8108e8ad33499b7356784559a49ae2c77"}]},"branch":"refs/heads/master"},"49140423c083de347c85abd637a804434c7b3211":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-08-31 23:01:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/7"}}},"commit":{"parents":[{"commit":"4170a10eb5d1d0239fb4858e40668aa70747222e","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4170a10eb5d1d0239fb4858e40668aa70747222e"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:01:28.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49140423c083de347c85abd637a804434c7b3211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49140423c083de347c85abd637a804434c7b3211"}]},"branch":"refs/heads/master"},"6160f7447f0f8b8fddc507dec68762d17441e37b":{"kind":"REWORK","_number":8,"created":"2025-08-31 23:06:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/8"}}},"commit":{"parents":[{"commit":"4170a10eb5d1d0239fb4858e40668aa70747222e","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4170a10eb5d1d0239fb4858e40668aa70747222e"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:05:29.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6160f7447f0f8b8fddc507dec68762d17441e37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6160f7447f0f8b8fddc507dec68762d17441e37b"}]},"branch":"refs/heads/master"},"0c582d3172272c24305428c37ca0a517f67200f4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-01 19:56:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/9"}}},"commit":{"parents":[{"commit":"d0164d3b4be75cae36154b315cb2c774cdfe6575","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0164d3b4be75cae36154b315cb2c774cdfe6575"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 19:56:33.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c582d3172272c24305428c37ca0a517f67200f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c582d3172272c24305428c37ca0a517f67200f4"}]},"branch":"refs/heads/master"},"5419f7dfa8ddf94d86ab187e662024e4f5e1c9bf":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-09-01 20:16:02.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/10"}}},"commit":{"parents":[{"commit":"a388148be6acdf402c093912f8855d3dad646903","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a388148be6acdf402c093912f8855d3dad646903"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 20:15:57.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5419f7dfa8ddf94d86ab187e662024e4f5e1c9bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5419f7dfa8ddf94d86ab187e662024e4f5e1c9bf"}]},"branch":"refs/heads/master"},"4d426daf96c4f59072703fd6c4b63ce2a7f245ce":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-09-01 22:02:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/11"}}},"commit":{"parents":[{"commit":"bf8f40fc0d74bc5bb5ce6fa4769bb55178fb52af","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bf8f40fc0d74bc5bb5ce6fa4769bb55178fb52af"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 22:01:43.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d426daf96c4f59072703fd6c4b63ce2a7f245ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d426daf96c4f59072703fd6c4b63ce2a7f245ce"}]},"branch":"refs/heads/master"},"1bdbf502d9e83ad069e3489be263c18fbc052e92":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-09-02 19:02:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/12"}}},"commit":{"parents":[{"commit":"6fa476deb9a154f27ec7aedf54350814f2d308e9","subject":"Update tests that were subject to race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6fa476deb9a154f27ec7aedf54350814f2d308e9"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 19:02:36.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bdbf502d9e83ad069e3489be263c18fbc052e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bdbf502d9e83ad069e3489be263c18fbc052e92"}]},"branch":"refs/heads/master"},"3798049e919ef75da9ddafbe947eef15c6725a22":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-09-02 20:43:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/13"}}},"commit":{"parents":[{"commit":"1340166ca170e3d582d498a3d5d00ad3fed56804","subject":"Update tests that were subject to race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1340166ca170e3d582d498a3d5d00ad3fed56804"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 20:43:13.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3798049e919ef75da9ddafbe947eef15c6725a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3798049e919ef75da9ddafbe947eef15c6725a22"}]},"branch":"refs/heads/master"},"e0ec0dc8641d03b0268d315fad1ba20a28a0a338":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-09-02 21:00:37.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/14"}}},"commit":{"parents":[{"commit":"599a2722a42a466ad1670d78b4a77558f4c2a885","subject":"Update tests that were subject to race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/599a2722a42a466ad1670d78b4a77558f4c2a885"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 21:00:31.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0ec0dc8641d03b0268d315fad1ba20a28a0a338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0ec0dc8641d03b0268d315fad1ba20a28a0a338"}]},"branch":"refs/heads/master"},"ae43e217151d3dd7b49b4cb616896fc0231f2cf6":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-09-03 19:13:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/15"}}},"commit":{"parents":[{"commit":"073a220cfaaf387c08c67c042073d66f50b32a94","subject":"Disable python unit tests temporarily","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/073a220cfaaf387c08c67c042073d66f50b32a94"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:08:18.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae43e217151d3dd7b49b4cb616896fc0231f2cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae43e217151d3dd7b49b4cb616896fc0231f2cf6"}]},"branch":"refs/heads/master"},"58690dfd7d54b4da4663f51979cb72a3278c31ad":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-09-03 22:04:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958595/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958595/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958595/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958595/16"}}},"commit":{"parents":[{"commit":"70c9e869360a0e836c098f021fe92fa7d39e9ad1","subject":"Disable python unit tests temporarily","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/70c9e869360a0e836c098f021fe92fa7d39e9ad1"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-26 22:12:58.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 22:03:31.000000000","tz":120},"subject":"Replace eventlet wsgi by cheroot","message":"Replace eventlet wsgi by cheroot\n\nLike ironic did previously, let\u0027s get rid of eventlet based wsgi and\nuse cheroot instead.\n\nAlso stop using oslo.service options, because they are now in conflict\nwith oslo_config_glue from cotyledon.\n\nChange-Id: I55172351d1768bdd34fab50bac228928ce7b1765\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58690dfd7d54b4da4663f51979cb72a3278c31ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58690dfd7d54b4da4663f51979cb72a3278c31ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
