)]}'
{"id":"openstack%2Fmistral~958705","triplet_id":"openstack%2Fmistral~master~Ia3869dec4f4fba6c090050125eee1df17e188427","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-04 17:36:27.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ia3869dec4f4fba6c090050125eee1df17e188427","subject":"Deprecate api_workers config option","status":"MERGED","created":"2025-08-27 21:53:06.000000000","updated":"2025-09-04 19:12:17.000000000","submitted":"2025-09-04 19:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958705-eventlet-removal","meta_rev_id":"1f10a5cb078dd2a5c896508440105953785332ad","_number":958705,"virtual_id_number":958705,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 19:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 08:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 08:50:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-27 22:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c73fbf77716b42798834819157cde14a2368df79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-27 21:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2d2bc1d9aa16194184c8733a70543f554c3e4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 22:46:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/364a19fbb17b4c94b3042f1c4e3a5064\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3194a1ba1684aea873fcb46d9514b4f : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/452dba56699c434b8d16f035e5c76c8e : TIMED_OUT in 41m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9f7155b1f054d1e953fd133092d7c5e : TIMED_OUT in 41m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3b819deec6147c098766aad5f04c71a : TIMED_OUT in 41m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e4bb0c09c344ce89cc15725094932e : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f757af5c1b5340b09084848dd7096276 : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/285a23c911d3460eb66041c4735cebcc : TIMED_OUT in 51m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c995491777d24aa58415cde2b8e3d053 : FAILURE in 29m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/113a2ca139aa4cd6a0fb2df8f72cde3b : FAILURE in 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"268869bab9dd0af001fd8d331a3737a5c4965236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-28 11:35:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ff3700c8391b037e0529bd951ed81af0d9c83ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:29:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2de0ce37abca439bb56ab67a756fd44f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0656202afdb8422594136b1e899853ed : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16a7600bbe8e4eecbaec907106af1814 : TIMED_OUT in 46m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1628241fb60949c786e0f842fcbcdf0a : TIMED_OUT in 40m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3aa62594562641ad85597b5169d46d61 : TIMED_OUT in 41m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f905c399541243df82b9cc014aef8844 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faed3528b91b4d43ae7b95fe88b858da : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df90f92166f84bf58f02e97d938525e9 : TIMED_OUT in 51m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/efa3c17f1cd84723a756fd3fa7126b04 : SUCCESS in 27m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3967dc5883ed49c4ba714cac248cca41 : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5debe45b38c33229736d8095bf36b0cb0ad3ae2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 19:12:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"62fadb3ccd33bde3eb92ac1cb90ffe0d5891a54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 20:08:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e51bc62fb2ce40e08becf63f4ef910fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f109f475a8234a78b7be172770d6a516 : FAILURE in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07d6134c3bfc448cbd3918043718fd8e : TIMED_OUT in 41m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13b256ba1888416d833e330246c2ce81 : TIMED_OUT in 40m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b42e66e6e7764daf91ab8c8b70837724 : TIMED_OUT in 40m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e79456f64e4e8e9e8b3d9a8e14346b : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c46df34c82f947238ec5b2aa9aec1c28 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46bc54c7a8d14c1685999cdcf545f1d0 : TIMED_OUT in 50m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b2027140b84c45d39118624f8cd5f221 : SUCCESS in 27m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce0851de19d446759ddb3ad265f8f08d : SUCCESS in 41m 20s\n\nWarning:\n  Comments left for invalid file mistral/db/v2/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":3},{"id":"f5c5df5607d4cf17fe596688dafface3564e96ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 22:50:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"169d24496121e1c00335b55cf4002fd391b94b93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:01:32.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ac46c1110f5ee3fa611ba9836aed578c9dd91224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:06:23.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e43543640b463c173ab2b9b7a219d8769a044aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 23:58:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9532084c97e840e9bda50f26976943b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880484199edc46a98c37d061c17ce9ce : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/891c3b3eddca42db894388b5df7b768b : TIMED_OUT in 40m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9237fef957441af998ca60c62268332 : TIMED_OUT in 40m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47eb9daf040447ccac1ebfb9cc6634a2 : TIMED_OUT in 40m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34af796b93c4190bd04355b89de3ba3 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6a6522f4fcb41ef84910ee0a5bb78dd : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63a569bddaac44a6a5c8d7cf14704f26 : TIMED_OUT in 50m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/00d0b2827a9d4b129a18844d9eca89f6 : SUCCESS in 37m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37d7feebabaa436f895fa4a647cba6df : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":6},{"id":"7b11fc823d4ea929c3c0289ca98b4044efedc5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 19:56:46.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"70591d2eb46a229f712a1b795e35356c81ad0826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 20:16:02.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"8e836cdeca703968ba7c20fda29470938e586e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 20:58:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51ce939eff37451fb65d5d3f711d8d38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51af51314334edfa49faf11dccc5526 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fa019e202fb45449193bf607994594a : FAILURE in 12m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1dd625eab4a643b6b02ba8f11b456193 : FAILURE in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f81418f2728447708085f610385c52c8 : FAILURE in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e682148380849298956d75adb0edb95 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65517dbb64514008a9eeab097e61efcc : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbe2137a888f416dbb0f87dfc6d88941 : FAILURE in 14m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fcac5614f87f4275a5b454c111c83caa : SUCCESS in 17m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0dfa88277f4edd95a154d1f569e5e5 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":8},{"id":"238758dd1a6ee56ff52bfdc22b7d548f1b97d8e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 22:02:17.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"131351ed472668ec306f332d4a83feeab455f46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 22:50:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/500c07ddb64d4a1cbce4d90ad5fb7991\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/837839eec2d24586b8e9e383a1462dc7 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d375d4c3415047f588f2c9900e519c8c : FAILURE in 11m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a30ae4d9e214aceab771c167710fe07 : FAILURE in 12m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/658c2d3210c14f43afecdcb5d99f4b3f : FAILURE in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a20361b6134c8e9ca68340d1813a00 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0027ea199c5f47978fe8ec5c53cfed01 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/224e9da3fe1f4f6a98b2cda265356092 : FAILURE in 12m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2d76d0d1bf6f48f6aa6b015487d1e4bc : SUCCESS in 43m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23af7e5b7643483a8aa7ed1b3d513311 : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":9},{"id":"a1607c83ae4c9fe7b333a5514f774fe64fa2d766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 19:02:46.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"5e8855c31f7fe4ce53a9859e7784c2081e4ef642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 19:55:09.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63646cfd99944ac393ff0db9825dfd0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352a8232334a43e39c120f09cf4aba25 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5864ba9f5164df2910cc38ddb1da2fa : FAILURE in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9681ab54c3b54971b5de85dac0aafe51 : FAILURE in 11m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8f42e91e08e44db822a91712ca0ea16 : FAILURE in 15m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde76e80838942abb80e07f1ac2d9794 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e74303ab5124adb9a09a50f9858841a : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccb278aed379469f8c9c9fe003b7db44 : FAILURE in 17m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2ce1ca10f8934ccaa9a85a647dff4c58 : SUCCESS in 38m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/943f7410d6e44a6c90a2a6f4c7f6c4d1 : FAILURE in 43m 15s","accounts_in_message":[],"_revision_number":10},{"id":"3ce5bca4dfe9fffd6c55e37aa4d3e1cce4ab1832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 20:43:28.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"04bd33eedbb8f62624fd6db795690e266345d27a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 21:00:37.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"eef1e59a13ca01c8462fa9c7d820bebf29a3a633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:47:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13fdbdebe4e64814b7ec36655ca530dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c8999debf7467baa43f12523c7a99b : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7d0cb429e7e4012ab56f84de3600059 : SUCCESS in 10m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af148b92a8c6487e9e62ea25e3068644 : SUCCESS in 10m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dae027ac0364db19b13cd8575c9c204 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f54f800a564d2fa90b9b94b63db839 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/162c99186b33497ca2c8025fd914aafe : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fe4c7c300b6409cb11ee8f50ab7c850 : FAILURE in 16m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5f1d04cd961a43e2937270c462bbc5c9 : SUCCESS in 37m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8292a3cfc4ca4bd9ab17016a7d0d8034 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":12},{"id":"35440143bd886e3f08fd62b5db600c5e8b55b771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 19:13:10.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"b7b596def9872c6d222c8a94246d0c4d017acb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 19:49:07.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6adffc1703f484c9d4af0f1bb9155d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1fdc476555b45569f4f3c291b60daca : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a9378547d194e50984ff2b858a0f364 : SUCCESS in 11m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4ed5afd3e064660a9fae4f166aa4722 : SUCCESS in 9m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ed700d25e0c43b48d3e44a722f5949d : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad93d0be161148f1a22bd12b70108b81 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8f6553320334e128a31d9ec1a6263f7 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80ec04b7c50140148cf2e55763aced2c : SUCCESS in 13m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/91916b1f486740b6bb14431c042ece9e : SUCCESS in 28m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb1f3247fb92408b8b1aaba1e799724c : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":13},{"id":"3c684ba88d70d29f3eda7cbd605b3e6b5e5660ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 20:12:20.000000000","message":"Uploaded patch set 14: New patch set was added with same tree, parent tree, and commit message as Patch Set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"25689da494099a6a4ad53d6a916599dff63fa6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 21:04:14.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a21e41579846858b6f9be6d2a191b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19434ffd62e8472491266a4da751f870 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74db7928116644389c98e06cd3f64a5f : SUCCESS in 13m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f46ee4804bf4550aaa0c224d169ca94 : SUCCESS in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d63c6891f4de4e0eb4103a6c188da79d : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffaad0e051e34550aae896f161755543 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e9707d061c44e1eab7f8cf42775fe00 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3ba4fa7e55d40adaa9bd521e0798315 : SUCCESS in 18m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ca571d58cd2945e0b8a8af75f73f63bb : SUCCESS in 41m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81e768bae06b4ee6a0fb429c95ce226d : SUCCESS in 25m 51s","accounts_in_message":[],"_revision_number":14},{"id":"ed81d00cb8b1badb3fef5af418679f38663bd003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 22:04:13.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"72955b0b834a8eecc09efc8562ac6f65ad7e82a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:46:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09da90669a404c73acdad5661b8f90b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75503bfba2334a2bb8bb7c53d72be66c : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40f3b53746754192af09d1969558f881 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28be3290e34240a8b4f70ce48256bab3 : SUCCESS in 10m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/498cf9ac72024b6897c717ac74952098 : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fcf8a1a2d34125b29b6929912e40be : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/550361934a0d48be92eb2830f72cbc61 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01224bca50584b56b3ce794523e601ef : SUCCESS in 16m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/236d1623555c491198214bdfe15d4b60 : SUCCESS in 16m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02405c9b5d4444bbb50d3c59a8d3bbc6 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":15},{"id":"3a331ec766885f92598e0a4fb2a54eb0da7c4586","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 08:50:53.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"c7d51ceb465e3d8a063e89e31e9b8405adc4fcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:51:47.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"49fec18e5c087785efaa177ecc6dcca04e1169eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:46.000000000","message":"Patch Set 15: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/959376 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"dccda3fed349f512558c91cd280306e99c66f09a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 17:36:27.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"47c3e2744a6427e97922cd817f2199a25c9c3d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:28:27.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd602af0a4a4b799fe045324ea4abe6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e543d388e34a8493e966c0053b162a : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b0de976b58845efbc45104cf7a44bda : SUCCESS in 11m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b58ceaf848eb48949d8542c69f040e9d : SUCCESS in 13m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6b9bb402c84459e95d8d9fa036ed341 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2292b6168e4050a186de51e03ef367 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e87c542a5f1649968f03d52ef9af25f3 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f54b829264a646c89968e603d1bb1cb3 : SUCCESS in 16m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/581caf1c9c9d49d08b3f7f18c0d297d9 : SUCCESS in 34m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fbf79a26d5a476ca7d03970b092e244 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":15},{"id":"7d65460d489314d2da19c19ac1d3880acf96a19a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:29:02.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"c39098e27e36b7bb5bae20ea1fca4c5160ea6c20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:10:51.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1160331eeaca493389c3e9b3797a307f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c19265050524cbaa94d890918be9460 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/784fc9df67844b228eb7d97ec4de4dde : SUCCESS in 11m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a26935db264241c18481af875676054a : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc7e2bbf98b4e498ec4643034707dd4 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb881b8fdd64c2aa2d47c407c40d9af : SUCCESS in 4m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9cde5a5c97df4b638293c8238f26c1b9 : SUCCESS in 40m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37caf35043884788bc6d95216f6e06c9 : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":15},{"id":"2f9c86d0fd9bbf276563eaca0d7516ab762e059d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"1f10a5cb078dd2a5c896508440105953785332ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:12:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5840481d5ee4969b2d31ea87e5d6c42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527a05c1000749c59772510d8ccab31f : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ec02173b3d843c583f9a30202d916a3 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"f7dda97d404b84bf7ba24ee517f890f5c41554bf","revisions":{"e56bb7f63e13e8a2cbdae3b218d7cdfca4fe592e":{"kind":"REWORK","_number":1,"created":"2025-08-27 21:53:06.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/1"}}},"commit":{"parents":[{"commit":"e9d52ca794afd0b4d69e87bbfee05b33a2077078","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9d52ca794afd0b4d69e87bbfee05b33a2077078"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e56bb7f63e13e8a2cbdae3b218d7cdfca4fe592e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e56bb7f63e13e8a2cbdae3b218d7cdfca4fe592e"}]},"branch":"refs/heads/master"},"f5640aa7c3c4804ff8c12d54eddfb25c2e313b52":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-28 11:35:42.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/2"}}},"commit":{"parents":[{"commit":"df549b3625d5908e99c1cd6d5b1c534929d363b1","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df549b3625d5908e99c1cd6d5b1c534929d363b1"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 11:35:29.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f5640aa7c3c4804ff8c12d54eddfb25c2e313b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f5640aa7c3c4804ff8c12d54eddfb25c2e313b52"}]},"branch":"refs/heads/master"},"592718c5e3105a92f99c7f9ba586887a29fdb8f0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-31 19:12:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/3"}}},"commit":{"parents":[{"commit":"6ea3654cac754947a206fc1c603a69f7773a5f90","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ea3654cac754947a206fc1c603a69f7773a5f90"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:12:10.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/592718c5e3105a92f99c7f9ba586887a29fdb8f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/592718c5e3105a92f99c7f9ba586887a29fdb8f0"}]},"branch":"refs/heads/master"},"5ab7309254cf40a693f793a897cdb5fbf1252c49":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-31 22:50:05.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/4"}}},"commit":{"parents":[{"commit":"3d476ef8108e8ad33499b7356784559a49ae2c77","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d476ef8108e8ad33499b7356784559a49ae2c77"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 22:49:02.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ab7309254cf40a693f793a897cdb5fbf1252c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ab7309254cf40a693f793a897cdb5fbf1252c49"}]},"branch":"refs/heads/master"},"4afefe9cccb4798b493fdbabcbc23fd80023401d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-31 23:01:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/5"}}},"commit":{"parents":[{"commit":"49140423c083de347c85abd637a804434c7b3211","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49140423c083de347c85abd637a804434c7b3211"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:01:28.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4afefe9cccb4798b493fdbabcbc23fd80023401d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4afefe9cccb4798b493fdbabcbc23fd80023401d"}]},"branch":"refs/heads/master"},"0826cb2b1adc0f3edeffdd2d9b9cd873d6054fad":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-08-31 23:06:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/6"}}},"commit":{"parents":[{"commit":"6160f7447f0f8b8fddc507dec68762d17441e37b","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6160f7447f0f8b8fddc507dec68762d17441e37b"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:05:29.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0826cb2b1adc0f3edeffdd2d9b9cd873d6054fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0826cb2b1adc0f3edeffdd2d9b9cd873d6054fad"}]},"branch":"refs/heads/master"},"bd053a932faa9a4299fd4137ed0f42ebdd03314f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-01 19:56:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/7"}}},"commit":{"parents":[{"commit":"0c582d3172272c24305428c37ca0a517f67200f4","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c582d3172272c24305428c37ca0a517f67200f4"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 19:56:33.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bd053a932faa9a4299fd4137ed0f42ebdd03314f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bd053a932faa9a4299fd4137ed0f42ebdd03314f"}]},"branch":"refs/heads/master"},"9a3eb786ae5efb4bdbac192783a2b23c677658bd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-09-01 20:16:02.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/8"}}},"commit":{"parents":[{"commit":"5419f7dfa8ddf94d86ab187e662024e4f5e1c9bf","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5419f7dfa8ddf94d86ab187e662024e4f5e1c9bf"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 20:15:57.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a3eb786ae5efb4bdbac192783a2b23c677658bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a3eb786ae5efb4bdbac192783a2b23c677658bd"}]},"branch":"refs/heads/master"},"882936202c814b099feb08c2555ff3ce0a3b5825":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-01 22:02:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/9"}}},"commit":{"parents":[{"commit":"4d426daf96c4f59072703fd6c4b63ce2a7f245ce","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d426daf96c4f59072703fd6c4b63ce2a7f245ce"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 22:01:43.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/882936202c814b099feb08c2555ff3ce0a3b5825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/882936202c814b099feb08c2555ff3ce0a3b5825"}]},"branch":"refs/heads/master"},"e3754800d07c0e703692f5f0e44a24713a7cef57":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-09-02 19:02:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/10"}}},"commit":{"parents":[{"commit":"e6387022ff1cd91905a247bba56ab715afef58be","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e6387022ff1cd91905a247bba56ab715afef58be"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 19:02:36.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3754800d07c0e703692f5f0e44a24713a7cef57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3754800d07c0e703692f5f0e44a24713a7cef57"}]},"branch":"refs/heads/master"},"5272b55daab545c590397dbce62d94edac012d03":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-09-02 20:43:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/11"}}},"commit":{"parents":[{"commit":"f76042ab3211bf788c98b6e703aef14ba7b330c3","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f76042ab3211bf788c98b6e703aef14ba7b330c3"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 20:43:13.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5272b55daab545c590397dbce62d94edac012d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5272b55daab545c590397dbce62d94edac012d03"}]},"branch":"refs/heads/master"},"042cdaf951bbac14465a5f02ccaac3ae8f873de3":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-09-02 21:00:37.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/12"}}},"commit":{"parents":[{"commit":"bbd87748b8fabbfdc9621ac0723e788fd2a42a47","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbd87748b8fabbfdc9621ac0723e788fd2a42a47"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 21:00:31.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/042cdaf951bbac14465a5f02ccaac3ae8f873de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/042cdaf951bbac14465a5f02ccaac3ae8f873de3"}]},"branch":"refs/heads/master"},"186ea7447e47c943cc68ca30ac577c64a94072a2":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-09-03 19:13:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/13"}}},"commit":{"parents":[{"commit":"f784c9aef0c791586910a0d94606f0e767b27087","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f784c9aef0c791586910a0d94606f0e767b27087"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:12:58.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/186ea7447e47c943cc68ca30ac577c64a94072a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/186ea7447e47c943cc68ca30ac577c64a94072a2"}]},"branch":"refs/heads/master"},"ea167170ee3957e3312d47f81fd9f037beb48e64":{"kind":"NO_CHANGE","_number":14,"created":"2025-09-03 20:12:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/14"}}},"commit":{"parents":[{"commit":"485dd1b40fdceabfac258fd8a03528cd848f3d37","subject":"Revert \"Disable python unit tests temporarily\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/485dd1b40fdceabfac258fd8a03528cd848f3d37"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 20:12:10.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea167170ee3957e3312d47f81fd9f037beb48e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea167170ee3957e3312d47f81fd9f037beb48e64"}]},"branch":"refs/heads/master"},"f7dda97d404b84bf7ba24ee517f890f5c41554bf":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-09-03 22:04:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/05/958705/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/958705/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/958705/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/958705/15"}}},"commit":{"parents":[{"commit":"2ec6f9c0a79c5e33e200baadc6e849e507e897f5","subject":"Revert \"Disable python unit tests temporarily\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ec6f9c0a79c5e33e200baadc6e849e507e897f5"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-27 21:52:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 22:03:31.000000000","tz":120},"subject":"Deprecate api_workers config option","message":"Deprecate api_workers config option\n\nSince we moved to cheroot, we can\u0027t start more than one worker.\n\nChange-Id: Ia3869dec4f4fba6c090050125eee1df17e188427\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7dda97d404b84bf7ba24ee517f890f5c41554bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7dda97d404b84bf7ba24ee517f890f5c41554bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
