)]}'
{"id":"openstack%2Fmistral~958995","triplet_id":"openstack%2Fmistral~master~I722b9f1ff091bef7df282afadd15cd3b3872d040","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-04 17:36:34.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I722b9f1ff091bef7df282afadd15cd3b3872d040","subject":"Get rid of oslo_rpc_executor option","status":"MERGED","created":"2025-08-31 19:12:32.000000000","updated":"2025-09-04 19:17:18.000000000","submitted":"2025-09-04 19:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958995-eventlet-removal","meta_rev_id":"b324bae0a1531122afe09da55ee0222abe5f74f0","_number":958995,"virtual_id_number":958995,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 19:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 08:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 08:51:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-31 20:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63d7f25389e156bd21e873c4684e2f59d32a6c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 19:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdf5177969966f13f8d02d5b42430e9a10e93e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 20:11:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f71df186a8644dd87602dd8fda0499e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e864df7e8d8648998f89d0528b5f7d6d : FAILURE in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31df2806d8ef4e45aa24b2c47003a651 : TIMED_OUT in 40m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7706369715744c11b76196de244fde82 : TIMED_OUT in 40m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a26e2ed041e48d2862b0f2dd95d58bc : TIMED_OUT in 40m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474b174f32354104a735a2bdc24a4276 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/392d4b8786a84460a6c9ca390206ee01 : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3ced09efb534b52b554cc78ecb8d3bc : TIMED_OUT in 50m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ef40a9fd1ad84803b4bd5930960a552c : SUCCESS in 39m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd5d7675d694ed6baa228ac15dc3281 : SUCCESS in 37m 53s\n\nWarning:\n  Comments left for invalid file mistral/db/v2/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"6392afbcd6ba24539e45f2d9ae6f8e2b3888e3c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 22:50:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c906d8ed37f17897ebfe7399a74b1bee0530e490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:01:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2ea645661e0d5aba6ea16267206ff170b4ec53a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:06:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a7efc8053eff4f7646197a9c611433056aeb9711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 00:02:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0897d00070f447ff96a2ea2225da7570\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64439153334a46b4a25b5210e66418f8 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8ec666530e14a6ea045571764a7e7a6 : TIMED_OUT in 40m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c724b3f74c124d899d3d69643e97433e : TIMED_OUT in 40m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14d8c9c928ed468b8aa60ce880267247 : TIMED_OUT in 40m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d79c504e3a4e2b9e1548ae4a453cf7 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0369eaf6152549a1987aefb84ece4c13 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9358fcc2c86946b8b0cfe7430947839b : TIMED_OUT in 50m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/95196f3a96e847d0aaae1eb073b871ea : SUCCESS in 20m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/460a40079e334b8ba2dd6c210ada6d46 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":4},{"id":"e25b53361093af17bc3e7fade1e3f5b96d0cae96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 19:56:46.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ec5be288238efdc5215b07767f6a4e2c54747bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 20:16:02.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"45bb05582085eb9f0545b1227905bddd6fa9bc89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 21:01:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b3f40c4ff454e26ac7866692bb9571f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f53181872c4eb1882d1f4fdbd4050e : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab4d91784a3449fa9b82f561c0b9b221 : FAILURE in 12m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd8cb1a727cb4777b57b4a680df22a97 : FAILURE in 13m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbbe41106d7d4dec8b9a63d2f59dbfbd : FAILURE in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cdb0cf30714c0390e6742965ad38a3 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ef9739a09624270a33c8fa341a1f0cd : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc4fb0fba8c449eab7887836006542ea : FAILURE in 16m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e09c081d954148ddb6b2813861db45dc : SUCCESS in 24m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a6be2f7743d4f8b9fca65d1ee52f88f : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":6},{"id":"107d087a65250106beb0e8a50e08e42f86f3f47d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 22:02:17.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"62adc8b7cd1f419879d413d34685416d92b59fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 22:47:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58d4dc51ed2e4d48b21c542c9c667103\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f700448f72fd466cb3e5f7ed0e33c22f : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ac4307b4c9e4df7a8fffbe6a3f579cb : FAILURE in 12m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bff738eba25c4fa798cbb54de6e8da4f : FAILURE in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ceca6a172c0c43b2bfe4fcbf89d78a59 : FAILURE in 15m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a8ca0ffdf44c829f36eccf308744ec : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e759cf1178ae42339d5dfe524340a93d : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0959181e7a64222a800b5265df0b785 : FAILURE in 16m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d3c820b3e2454b6d92e38c8f368fd9d6 : SUCCESS in 16m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28b3be60ad484447b11fe5e37ad8b8be : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":7},{"id":"9371acbfb4212bd0f6a46f64d0c4e595ca15cd1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 19:02:46.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"514cc4697e96c40e071a2a32b79e861b7ea51ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 19:47:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0443e422f4024c55926d4a5c79046d77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529042527e0144e4b1ed8d98734ea6a2 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af0f8b3e854e4bd4ab2f9427f06f9540 : FAILURE in 13m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46b6b75be80f439c98b792ce0bd611a5 : FAILURE in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d325918f8339443bb21ab164e52e52b2 : FAILURE in 15m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85dd8ad58c7464fb9bdcd05217ceaf6 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/351273fc42024e6bb0cf44bc71de8437 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f85d10ef70fc4ba2803d41f1ec08d168 : FAILURE in 15m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e215681e0bb6420b9871db8ad4ed542b : SUCCESS in 16m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ab49dadd7164400bdbd15e716fb5048 : FAILURE in 37m 06s","accounts_in_message":[],"_revision_number":8},{"id":"f05e1a0ff779e0e95d0dd3041d1685077149ed07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 20:43:28.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"dbf8d94985aa8964fdab748ae145c98a7d50e11f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-02 21:00:37.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"d14434cb06fa85e5fcd1eb4bbf8c060b39d9c226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:46:57.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d337330cf6624781bf5c09b4b5811e49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4d41ffd9c340da9f7ecae13429c8b1 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc8d1a7978b443ee81d1789311265b50 : SUCCESS in 11m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/623ade8f5f95446e9c67fa5b28ca3c92 : SUCCESS in 11m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00f1ce1df9174500ac49b6327508135b : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6de6578dec4e87ac5d897a5adf072d : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e673b7f3a57d4c6faec0141d4e3e5256 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a773f8ddd7c647f4baff158fa0a0f1df : FAILURE in 16m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2792a639367946c899ff668b93e9bad1 : SUCCESS in 39m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dfb821aeb4b461cb3cf3066c27fe6cb : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":10},{"id":"331ee1be1d7bf1f3236197dbc0dbad46074ffa9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 19:13:10.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"7f1b1568bb7988c034f625d2d37b543af3c03456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:04:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7edd48624c8e4f0da7fa00d1372d22d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f59d933a2545ee8fcefa428cdd204a : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71991d74fcd34575999c03af7e870681 : SUCCESS in 12m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0ef458dbec84ce3bb4682e300d1403b : SUCCESS in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79c8f2f4ff014f22bb996413055a8f0b : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe874fc8d224847acf16ce859bcd5de : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caa0d58c5ca24e3a8afd993fb123c53f : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8e08939eb714a67bbf40e4285a27488 : SUCCESS in 12m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/12328dcb4ff54e92b0867bcc0cb90551 : SUCCESS in 41m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4507d99bca7f4cf68ebfa9f4103328b6 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":11},{"id":"1df34521b6fd6c7165c16745e6b56934a82a7f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 20:12:20.000000000","message":"Uploaded patch set 12: New patch set was added with same tree, parent tree, and commit message as Patch Set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"c13a11469e79fb0300377e7c3b3153d487758a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:59:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b71f437c5a70418d9a061d381cdfdc83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c9112d15a84cef8fccd749dcd46a1a : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b70f8fdb538f4871a8861645063da28e : FAILURE in 8m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33c57d2460ee4d469eaf31e5ce182b94 : SUCCESS in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3de9f4d7183d49bdbe5898b11e71f26c : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f7095492e24133a61cbab0c5335f1e : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f3003a711e14667bf7c006b12320faa : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e34c2c4aef0443bb7d00c7fda3fbb7d : SUCCESS in 17m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e243118dc54745b8b492a146bfe81ef9 : SUCCESS in 37m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fa24cec7b6149a9ba0b13ef6bbc2eac : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":12},{"id":"99dd8e3a4941b4ede2e87e641e7e79e9261ebb69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 22:04:13.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"54154de2012267bfec1928a95d0ab28e7b2a9ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:51:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e4365297eee484ba81164264b35abc1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2856e45def524cc39360ee712330bcbf : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69ab7164878b41f3904f852d40c4345e : SUCCESS in 12m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64042b57c5724833b7a57de3c29c4f80 : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/595769ef220b446e9db92696598cd991 : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce762ea6a6f47e199ea728c199a981c : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc18e18c265142d3abad57bf276f4ee2 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ad895e6782f495cac467e4456d6927f : SUCCESS in 18m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5c0787efae894ed4af6a7aadf50171c4 : SUCCESS in 37m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0efcc2fe7ef4380a68030f8f0fc7a16 : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":13},{"id":"e22b6a13d3426a66020820a366bf074a54b8c80f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 08:51:20.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"21a5cf85a37d81d5ce4e2bece4600ea928088af8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:52:29.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"87645d16cd749d29f8395242021b41333dfcb37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:47.000000000","message":"Patch Set 13: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/958705 is needed.","accounts_in_message":[],"_revision_number":13},{"id":"035958505b1b3abccc538c6bbb3f74bb65883c7c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 17:36:34.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6ff72cac38bdc89da6bff16ec31488264b7ea22e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:18:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d1069f5fe4549b2ac9dd1994e2412a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3292384caaea4b07beff263691f2ee90 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f38b33ff74174b26bd199ee9c95fe1dc : SUCCESS in 9m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/964de6a7fdd342a7afd0338f0dc7ed47 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73c8268058ab4731b9f714eb4c491f77 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4edec92ca34026b0d50f7bea4458a1 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7fd963dccb04391a567c41f2960f01f : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acfcf316e1fe4c76a78735e4c02019e5 : SUCCESS in 14m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1306f9ad466e41b8b56cced7bd8ae887 : SUCCESS in 20m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98835e9a5d3f4dfa964219c39b2544bf : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":13},{"id":"92504294986a1d2f68848824ef3c872b6bc89c1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:29:29.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"245d35f092ffd34d83aa9fae010a32a051c4483c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:14:47.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdee7ee47c7e4b459f9e0761022d3656\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9049881d6b24775b818777df5a770e8 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb22605437a145d3b87a0bdfaeb6bde6 : SUCCESS in 10m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b79d87b726840fb84418dfbc126d1c6 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26380a0293f455c95b2939ea616e289 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec2dfa062cde4f108d6aa14aa5fa8c11 : SUCCESS in 3m 30s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/27d83b3c326241cbbf24951835a4904a : SUCCESS in 40m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7e332ba8e0d424b9664dd43ad5ff4c3 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":13},{"id":"6992c2b17c26a77ae4d6322d31a52cbd1cf8280b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:14:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"b324bae0a1531122afe09da55ee0222abe5f74f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:17:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818715499b854354a20f158195dea767\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c776e7aebc7149c2b78f9802b8406b59 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b74a25f0a8498495eab1f5ce18394f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"81c32801fff2130af0e96f3800ef2a654b3c42de","revisions":{"50c5cc3d0caa7891e981c28c86dcd922d166d422":{"kind":"REWORK","_number":1,"created":"2025-08-31 19:12:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/1"}}},"commit":{"parents":[{"commit":"592718c5e3105a92f99c7f9ba586887a29fdb8f0","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/592718c5e3105a92f99c7f9ba586887a29fdb8f0"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:12:10.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50c5cc3d0caa7891e981c28c86dcd922d166d422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50c5cc3d0caa7891e981c28c86dcd922d166d422"}]},"branch":"refs/heads/master"},"8b6d28042ef69f4958e9af4f1473e8a6ba61dde7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-31 22:50:05.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/2"}}},"commit":{"parents":[{"commit":"5ab7309254cf40a693f793a897cdb5fbf1252c49","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ab7309254cf40a693f793a897cdb5fbf1252c49"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 22:49:02.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8b6d28042ef69f4958e9af4f1473e8a6ba61dde7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8b6d28042ef69f4958e9af4f1473e8a6ba61dde7"}]},"branch":"refs/heads/master"},"6a01ec45bdc62f2604a3e80bceaa9ff542cc1222":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-31 23:01:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/3"}}},"commit":{"parents":[{"commit":"4afefe9cccb4798b493fdbabcbc23fd80023401d","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4afefe9cccb4798b493fdbabcbc23fd80023401d"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:01:28.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a01ec45bdc62f2604a3e80bceaa9ff542cc1222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a01ec45bdc62f2604a3e80bceaa9ff542cc1222"}]},"branch":"refs/heads/master"},"effec2bed2c9ea7b7843923d6e30ff41d0847857":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-31 23:06:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/4"}}},"commit":{"parents":[{"commit":"0826cb2b1adc0f3edeffdd2d9b9cd873d6054fad","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0826cb2b1adc0f3edeffdd2d9b9cd873d6054fad"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:05:29.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/effec2bed2c9ea7b7843923d6e30ff41d0847857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/effec2bed2c9ea7b7843923d6e30ff41d0847857"}]},"branch":"refs/heads/master"},"303a23cc3c04f6910fe5ea522375b6cccefae5bb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-01 19:56:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/5"}}},"commit":{"parents":[{"commit":"bd053a932faa9a4299fd4137ed0f42ebdd03314f","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bd053a932faa9a4299fd4137ed0f42ebdd03314f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 19:56:33.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/303a23cc3c04f6910fe5ea522375b6cccefae5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/303a23cc3c04f6910fe5ea522375b6cccefae5bb"}]},"branch":"refs/heads/master"},"c4251b2e17f92376acec823087706c4925d1b88b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-09-01 20:16:02.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/6"}}},"commit":{"parents":[{"commit":"9a3eb786ae5efb4bdbac192783a2b23c677658bd","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a3eb786ae5efb4bdbac192783a2b23c677658bd"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 20:15:57.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4251b2e17f92376acec823087706c4925d1b88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4251b2e17f92376acec823087706c4925d1b88b"}]},"branch":"refs/heads/master"},"6faa4f95610b45d2c076944fdd2b98c7417a9fbd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-01 22:02:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/7"}}},"commit":{"parents":[{"commit":"882936202c814b099feb08c2555ff3ce0a3b5825","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/882936202c814b099feb08c2555ff3ce0a3b5825"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 22:01:43.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6faa4f95610b45d2c076944fdd2b98c7417a9fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6faa4f95610b45d2c076944fdd2b98c7417a9fbd"}]},"branch":"refs/heads/master"},"99b768d152bd9d7203d30f2865f09e96a01dfbdb":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-09-02 19:02:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/8"}}},"commit":{"parents":[{"commit":"e3754800d07c0e703692f5f0e44a24713a7cef57","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3754800d07c0e703692f5f0e44a24713a7cef57"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 19:02:36.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99b768d152bd9d7203d30f2865f09e96a01dfbdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99b768d152bd9d7203d30f2865f09e96a01dfbdb"}]},"branch":"refs/heads/master"},"f9dd984a9da032251f9c7cb867d582199ac2ed3e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-02 20:43:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/9"}}},"commit":{"parents":[{"commit":"5272b55daab545c590397dbce62d94edac012d03","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5272b55daab545c590397dbce62d94edac012d03"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 20:43:13.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f9dd984a9da032251f9c7cb867d582199ac2ed3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f9dd984a9da032251f9c7cb867d582199ac2ed3e"}]},"branch":"refs/heads/master"},"741696c7ed36e17e934c7e903e435d815770f991":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-09-02 21:00:37.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/10"}}},"commit":{"parents":[{"commit":"042cdaf951bbac14465a5f02ccaac3ae8f873de3","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/042cdaf951bbac14465a5f02ccaac3ae8f873de3"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-02 21:00:31.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/741696c7ed36e17e934c7e903e435d815770f991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/741696c7ed36e17e934c7e903e435d815770f991"}]},"branch":"refs/heads/master"},"5c7ee4ef048a9a77d28aaa176532053c7eafc780":{"kind":"REWORK","_number":11,"created":"2025-09-03 19:13:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/11"}}},"commit":{"parents":[{"commit":"186ea7447e47c943cc68ca30ac577c64a94072a2","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/186ea7447e47c943cc68ca30ac577c64a94072a2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:12:58.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c7ee4ef048a9a77d28aaa176532053c7eafc780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c7ee4ef048a9a77d28aaa176532053c7eafc780"}]},"branch":"refs/heads/master"},"e1649b73085ed242999d36995dea86bdd10f21c6":{"kind":"NO_CHANGE","_number":12,"created":"2025-09-03 20:12:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/12"}}},"commit":{"parents":[{"commit":"ea167170ee3957e3312d47f81fd9f037beb48e64","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea167170ee3957e3312d47f81fd9f037beb48e64"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 20:12:10.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e1649b73085ed242999d36995dea86bdd10f21c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e1649b73085ed242999d36995dea86bdd10f21c6"}]},"branch":"refs/heads/master"},"81c32801fff2130af0e96f3800ef2a654b3c42de":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-09-03 22:04:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/958995/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/958995/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/958995/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/958995/13"}}},"commit":{"parents":[{"commit":"f7dda97d404b84bf7ba24ee517f890f5c41554bf","subject":"Deprecate api_workers config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7dda97d404b84bf7ba24ee517f890f5c41554bf"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-28 17:37:52.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 22:03:31.000000000","tz":120},"subject":"Get rid of oslo_rpc_executor option","message":"Get rid of oslo_rpc_executor option\n\nThe only valid option for engine rpc executor is now threading.\n\nChange-Id: I722b9f1ff091bef7df282afadd15cd3b3872d040\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/81c32801fff2130af0e96f3800ef2a654b3c42de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/81c32801fff2130af0e96f3800ef2a654b3c42de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
