)]}'
{"id":"openstack%2Fmistral~958996","triplet_id":"openstack%2Fmistral~master~I96202e201e9227fc0fb58780bec4aba20cc0e273","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-01 21:57:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I96202e201e9227fc0fb58780bec4aba20cc0e273","subject":"Add a warning about one test that is suspected to race","status":"ABANDONED","created":"2025-08-31 19:12:32.000000000","updated":"2025-09-01 21:57:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ea05d805cff2528add955cb72ddd020625e83c56","_number":958996,"virtual_id_number":958996,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-01 20:56:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-31 20:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d91abc6ff1acaef739017496881fdfe24b66e54d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 19:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c41cea5a8e8412fc550d323e9c8d0206d2da75bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 20:11:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/050893b7651a4c108fe162ed8965edcb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a114bfd9574e679dd5cc098fd14b77 : FAILURE in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b160757fa524970803a21cade32d0c9 : TIMED_OUT in 40m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4396eef9a2a2499da5ddb1ac9f39f303 : TIMED_OUT in 40m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff6ea7c4cdcc45bfbcfdc9e3ccee36f0 : TIMED_OUT in 40m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eafc1de1a7843a1bb5a699db027918f : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f02d37da7c914a4fa4ad24b2e51741a3 : TIMED_OUT in 50m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5d304e2c5f9b4d07b23cf3d6394685fb : SUCCESS in 22m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edcec188b9bd4bdab053797817d4f4cf : SUCCESS in 21m 50s\n\nWarning:\n  Comments left for invalid file mistral/db/v2/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"26335b2b10d1eccdc99f8847e527a6e6a0672ea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 22:50:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"75050d26cfd668316c043ef377c73a208ec6baa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:01:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b5c95eef16a7c67e7b0486c3f9d9b70bf8a0045c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-31 23:06:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e3cf621ac961d2ea21c70b07ad16c178ea0430c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 00:06:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5adc5c37cfb041b68cb40e35dad00f03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab2c5bf6fb54e86b5988ce6cf6ec967 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b01be109d2744d6af0fbd64a9b693ee : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/601760041d9449fe986d78f47cc4fa62 : TIMED_OUT in 40m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc722484d30f4e8c89cb3443e879e67f : TIMED_OUT in 40m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f9ad86c149b42c1bd0aae2af15b672c : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cebbde9ed6f7469fa10c3ffdf416f048 : TIMED_OUT in 51m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/58001233591b4b548ded252e23681852 : SUCCESS in 38m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d6e3d6c0bed43f4b492ebf0024ee35f : SUCCESS in 28m 58s","accounts_in_message":[],"_revision_number":4},{"id":"89319747a2b3df96c5bd0735b11bbbac34f5ce83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 19:56:46.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"471817eb0290e18777de6c1bd7161f2e835f5dbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 20:16:02.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"8b415c6192b7ab5259b46f11bd276dea026ad8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 20:56:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a3fd64afa9b4499bf797751b3d47a09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be246d78d6d4a8390441a1a620e48fb : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58832899757646909a73e7dd4a643094 : FAILURE in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/837004f2dd2b4e1298ebf367c23a41a5 : FAILURE in 10m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2efbc14c3bb4411ca1a2f8c7fb75e36d : FAILURE in 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec46ef2ee0584c1f9ebd10c4706e6d08 : SUCCESS in 8m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47fb8c43ad7e423aa66c21b44f0860b2 : FAILURE in 15m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4a83e3539ae84810a966fa39c6f0b92f : SUCCESS in 19m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6df795a56a474f3b895d77400d61c7c6 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":6},{"id":"ea05d805cff2528add955cb72ddd020625e83c56","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 21:57:03.000000000","message":"Abandoned\n\nI found the issue, it\u0027s because I was restart task t3 and it\u0027s not needed","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ea99212111861ccc5d9ba3920a79c914ab71f2c1","revisions":{"1f1f326d2fe76091d23a22f4351cb9e44d285f60":{"kind":"REWORK","_number":1,"created":"2025-08-31 19:12:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/958996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/958996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/958996/1"}}},"commit":{"parents":[{"commit":"50c5cc3d0caa7891e981c28c86dcd922d166d422","subject":"Get rid of oslo_rpc_executor option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50c5cc3d0caa7891e981c28c86dcd922d166d422"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:08:41.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:12:10.000000000","tz":120},"subject":"Add a warning about one test that is suspected to race","message":"Add a warning about one test that is suspected to race\n\nDuring my local debug session around threading stuff (eventlet removal),\nI noticed that at least one test could be failing due to race between\nthreads.\nI was not able to reproduce it after cleaning my code, so I\u0027ll bet on\nthe future.\nBut if my bet is wrong, we will have this comment that could be re-used\nas a starting point (or at least to try to refresh my mind about it).\n\nChange-Id: I96202e201e9227fc0fb58780bec4aba20cc0e273\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f1f326d2fe76091d23a22f4351cb9e44d285f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f1f326d2fe76091d23a22f4351cb9e44d285f60"}]},"branch":"refs/heads/master"},"537e9898f3a2eb98adaef26d6532a173162a315f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-31 22:50:05.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/958996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/958996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/958996/2"}}},"commit":{"parents":[{"commit":"8b6d28042ef69f4958e9af4f1473e8a6ba61dde7","subject":"Get rid of oslo_rpc_executor option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8b6d28042ef69f4958e9af4f1473e8a6ba61dde7"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:08:41.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 22:49:02.000000000","tz":120},"subject":"Add a warning about one test that is suspected to race","message":"Add a warning about one test that is suspected to race\n\nDuring my local debug session around threading stuff (eventlet removal),\nI noticed that at least one test could be failing due to race between\nthreads.\nI was not able to reproduce it after cleaning my code, so I\u0027ll bet on\nthe future.\nBut if my bet is wrong, we will have this comment that could be re-used\nas a starting point (or at least to try to refresh my mind about it).\n\nChange-Id: I96202e201e9227fc0fb58780bec4aba20cc0e273\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/537e9898f3a2eb98adaef26d6532a173162a315f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/537e9898f3a2eb98adaef26d6532a173162a315f"}]},"branch":"refs/heads/master"},"345b036f70db118e892ace16dac7c74abd3ffa2e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-31 23:01:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/958996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/958996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/958996/3"}}},"commit":{"parents":[{"commit":"6a01ec45bdc62f2604a3e80bceaa9ff542cc1222","subject":"Get rid of oslo_rpc_executor option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a01ec45bdc62f2604a3e80bceaa9ff542cc1222"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:08:41.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:01:28.000000000","tz":120},"subject":"Add a warning about one test that is suspected to race","message":"Add a warning about one test that is suspected to race\n\nDuring my local debug session around threading stuff (eventlet removal),\nI noticed that at least one test could be failing due to race between\nthreads.\nI was not able to reproduce it after cleaning my code, so I\u0027ll bet on\nthe future.\nBut if my bet is wrong, we will have this comment that could be re-used\nas a starting point (or at least to try to refresh my mind about it).\n\nChange-Id: I96202e201e9227fc0fb58780bec4aba20cc0e273\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/345b036f70db118e892ace16dac7c74abd3ffa2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/345b036f70db118e892ace16dac7c74abd3ffa2e"}]},"branch":"refs/heads/master"},"e98e8261f4635d1acaa5213f46db9517340a00f4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-31 23:06:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/958996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/958996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/958996/4"}}},"commit":{"parents":[{"commit":"effec2bed2c9ea7b7843923d6e30ff41d0847857","subject":"Get rid of oslo_rpc_executor option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/effec2bed2c9ea7b7843923d6e30ff41d0847857"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:08:41.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 23:05:29.000000000","tz":120},"subject":"Add a warning about one test that is suspected to race","message":"Add a warning about one test that is suspected to race\n\nDuring my local debug session around threading stuff (eventlet removal),\nI noticed that at least one test could be failing due to race between\nthreads.\nI was not able to reproduce it after cleaning my code, so I\u0027ll bet on\nthe future.\nBut if my bet is wrong, we will have this comment that could be re-used\nas a starting point (or at least to try to refresh my mind about it).\n\nChange-Id: I96202e201e9227fc0fb58780bec4aba20cc0e273\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e98e8261f4635d1acaa5213f46db9517340a00f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e98e8261f4635d1acaa5213f46db9517340a00f4"}]},"branch":"refs/heads/master"},"873651f14888d5257a8736f1e72702e23939e061":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-01 19:56:46.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/958996/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/958996/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/958996/5"}}},"commit":{"parents":[{"commit":"303a23cc3c04f6910fe5ea522375b6cccefae5bb","subject":"Get rid of oslo_rpc_executor option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/303a23cc3c04f6910fe5ea522375b6cccefae5bb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:08:41.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 19:56:33.000000000","tz":120},"subject":"Add a warning about one test that is suspected to race","message":"Add a warning about one test that is suspected to race\n\nDuring my local debug session around threading stuff (eventlet removal),\nI noticed that at least one test could be failing due to race between\nthreads.\nI was not able to reproduce it after cleaning my code, so I\u0027ll bet on\nthe future.\nBut if my bet is wrong, we will have this comment that could be re-used\nas a starting point (or at least to try to refresh my mind about it).\n\nChange-Id: I96202e201e9227fc0fb58780bec4aba20cc0e273\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/873651f14888d5257a8736f1e72702e23939e061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/873651f14888d5257a8736f1e72702e23939e061"}]},"branch":"refs/heads/master"},"ea99212111861ccc5d9ba3920a79c914ab71f2c1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-09-01 20:16:02.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/958996/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/958996/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/958996/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/958996/6"}}},"commit":{"parents":[{"commit":"c4251b2e17f92376acec823087706c4925d1b88b","subject":"Get rid of oslo_rpc_executor option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4251b2e17f92376acec823087706c4925d1b88b"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-08-31 19:08:41.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 20:15:57.000000000","tz":120},"subject":"Add a warning about one test that is suspected to race","message":"Add a warning about one test that is suspected to race\n\nDuring my local debug session around threading stuff (eventlet removal),\nI noticed that at least one test could be failing due to race between\nthreads.\nI was not able to reproduce it after cleaning my code, so I\u0027ll bet on\nthe future.\nBut if my bet is wrong, we will have this comment that could be re-used\nas a starting point (or at least to try to refresh my mind about it).\n\nChange-Id: I96202e201e9227fc0fb58780bec4aba20cc0e273\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea99212111861ccc5d9ba3920a79c914ab71f2c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea99212111861ccc5d9ba3920a79c914ab71f2c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
