)]}'
{"id":"openstack%2Fmistral~959099","triplet_id":"openstack%2Fmistral~master~Ia3b9d1760b17650cb9d8aa8cd74c1fb0c9dd6290","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Ia3b9d1760b17650cb9d8aa8cd74c1fb0c9dd6290","subject":"Remove a race condition in a test","status":"MERGED","created":"2025-09-01 22:02:17.000000000","updated":"2025-09-04 09:41:10.000000000","submitted":"2025-09-04 09:41:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959099","meta_rev_id":"13c997d5a452c5c099225e07ae9ad2a09f113ee7","_number":959099,"virtual_id_number":959099,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 09:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 08:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 08:47:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-01 22:49:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c978b552babfcf1fdcca7dd6dc9ec618f307135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-01 22:02:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"133a19750fdd0233693a074a1876cc95d14e6a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 22:49:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21f4705c354c45d7b33c961970f4a141\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b6194f241d467f9aaf65f42cd2f38c : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0381a630af8b4f2db590e9f7e82c8543 : SUCCESS in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4dbd73341b3f4776af82a21e0aac2d95 : SUCCESS in 13m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a847dd75f37442529fd55be9295b9b4e : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b1a08431c15411fb297a7def4ac6b21 : SUCCESS in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07158a09aef54d06a5c72edcbef15575 : SUCCESS in 15m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8cbee5f7f0414826bfa981f3a56f6d75 : SUCCESS in 37m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6a373902ee44f948db8c123c4292b21 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2c49329d5dc11daf3c86c1cb57e86fcdcae6fb47","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 08:47:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4cb7ddf2390ac93511d076d7c5325e060716db2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:48:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a22fc209c2411ab981a0aae2793aed069aec2dba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ea756316734fdb8a21966d6439c545\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf5328e5f774ccaa7764b807243ef6b : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/156594ce657049c686e386201f3b7282 : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d5cc5db625a4f218c185c5a699a6f7d : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631a89706e9c43dda44863346b753fd1 : SUCCESS in 6m 55s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3a2f8efcda20456891333df9626f8ef6 : SUCCESS in 27m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/445779a6ed3f4e52a2138dc9491c95a6 : SUCCESS in 30m 16s","accounts_in_message":[],"_revision_number":1},{"id":"13c997d5a452c5c099225e07ae9ad2a09f113ee7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3b8d22ce2618ea8dc58deaad96f39fd263d3ed6","revisions":{"c3b8d22ce2618ea8dc58deaad96f39fd263d3ed6":{"kind":"REWORK","_number":1,"created":"2025-09-01 22:02:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/99/959099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/959099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/959099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/959099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/959099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/959099/1"}}},"commit":{"parents":[{"commit":"c49116fbcbcbb9b38c9b5f62e5fb86217c7da51f","subject":"Use a threading.RLock for SQL sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c49116fbcbcbb9b38c9b5f62e5fb86217c7da51f"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 21:58:34.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-01 22:01:43.000000000","tz":120},"subject":"Remove a race condition in a test","message":"Remove a race condition in a test\n\nDuring my debug session, I had regular races condition in\ntest_rerun_join_with_branch_errors because we were restart task t3\nalongside with t1 and t2.\nIt t3 action was executed after t1 and t2, then it would be executed\ntwice:\n- from ERROR to RUNNING (and SUCCESS),\n- then again from SUCCESS to RUNNING (and SUCCESS again).\n\nThe correct behavior is to let it beeing started by t1 and t2.\n\nChange-Id: Ia3b9d1760b17650cb9d8aa8cd74c1fb0c9dd6290\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3b8d22ce2618ea8dc58deaad96f39fd263d3ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3b8d22ce2618ea8dc58deaad96f39fd263d3ed6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
